)]}'
{"id":"openstack%2Fproject-config~892650","triplet_id":"openstack%2Fproject-config~master~Ic30e7d39b641ef58d9a299c704be056e63c7a5d3","project":"openstack/project-config","branch":"master","topic":"charm-magnum-k8s","attention_set":{},"removed_from_attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2023-08-24 16:18:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic30e7d39b641ef58d9a299c704be056e63c7a5d3","subject":"Add OpenStack K8S Magnum charms","status":"MERGED","created":"2023-08-24 08:57:50.000000000","updated":"2023-08-24 16:33:26.000000000","submitted":"2023-08-24 16:18:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892650-charm-magnum-k8s","meta_rev_id":"f2409bed01180a3ff226c51c76cc074227097469","_number":892650,"virtual_id_number":892650,"owner":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 16:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2023-08-24 15:16:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-24 12:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-24 16:06:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 09:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 12:43:07.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-08-24 15:16:12.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"}],"messages":[{"id":"77503a176fd44f41db1ce04db392d07b8a8f4801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-08-24 08:57:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e574b7aae275887aa06f2f68c11dee71545202b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 09:12:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/601474dc62eb4344a005a98ab4503b19\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/98fb88a09ee0475f9326298ac41d2648 : SUCCESS in 6m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8c1cd895cba4477d926efb4371b53ace : FAILURE in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9d1f66db4e5748d2a81a19008a38a430a1235a2d","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-08-24 09:15:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25cb1dfaa6c852585f96fc0898cb306b58b5a07d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 09:24:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcbc92f1e92e4c8e8824f81114973acf\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bd2786655aca4c628f4fa410878fb8dd : SUCCESS in 6m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/87fd2b5512f94402bffac24923bd643e : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a9baf2a302bafd173e64fcc25d1c49f47efc56c7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-24 12:43:07.000000000","message":"Patch Set 1: Code-Review+2\n\nIt looks like I60fac7028ef9c8f20a81bdcedbc7740acec30b1b is the corresponding governance change. We\u0027ll need to see confirmation from the Sunbeam PTL either on this change or that one before we merge the project creation.","accounts_in_message":[],"_revision_number":1},{"id":"7f20119e9960de9da421104dd0fb23bbbae14f83","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-08-24 15:16:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9418ece3c4454c1622aaf56a5d97417139f3ebdd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-24 16:06:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11ab1b826c962a314469da4477fe13266a5e9637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:06:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77077edaa55671823df1257e4b8a7fdf34b0a66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:18:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71372eeb442048da9ea13a803240c756\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/87004ca1cac8472bb418dbf3ced6b305 : SUCCESS in 6m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/854c8b906c684c868bda17bdb0a2d257 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f0c3da432c27dbba3c12d302d721ee7e1a71d6d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:18:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2409bed01180a3ff226c51c76cc074227097469","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:33:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c784260e04464077b6f9d7ec08219a8d\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/fe1ba369b5044a54b12607817e5872ab : SUCCESS in 6m 27s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/2ca17e19f83543b7b15869ac00821744 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bedc2fad45fe904c92cf8dc329f4282c8e5d8115","revisions":{"bedc2fad45fe904c92cf8dc329f4282c8e5d8115":{"kind":"REWORK","_number":1,"created":"2023-08-24 08:57:50.000000000","uploader":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"ref":"refs/changes/50/892650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/892650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/892650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/892650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/892650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/892650/1"}}},"commit":{"parents":[{"commit":"d2a728b134106ed4c530bd3f37c79c09c8fa1017","subject":"Switch opendev tenant to Ansible 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2a728b134106ed4c530bd3f37c79c09c8fa1017"}]}],"author":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-08-24 08:57:14.000000000","tz":120},"committer":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-08-24 08:57:48.000000000","tz":120},"subject":"Add OpenStack K8S Magnum charms","message":"Add OpenStack K8S Magnum charms\n\nSetup and import OpenStack K8S Magnum Charm for deployment of\nmagnum hosted in Kubernetes.\n\nChange-Id: Ic30e7d39b641ef58d9a299c704be056e63c7a5d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bedc2fad45fe904c92cf8dc329f4282c8e5d8115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bedc2fad45fe904c92cf8dc329f4282c8e5d8115"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
