)]}'
{"id":"openstack%2Fproject-config~893703","triplet_id":"openstack%2Fproject-config~master~I68ea25b380eda353f9e0ad0868bdf7f6730a6733","project":"openstack/project-config","branch":"master","topic":"update-regexes","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-05 07:04:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I68ea25b380eda353f9e0ad0868bdf7f6730a6733","subject":"DNM: testing zuul config error","status":"ABANDONED","created":"2023-09-05 06:51:17.000000000","updated":"2023-09-05 07:04:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb9cc771a0101ec80e7de5c74b6ce355fab98bc2","_number":893703,"virtual_id_number":893703,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-05 07:02:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 06:51:48.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-05 06:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8aca72002463f4eab53a097c1cf4a1af5aff63e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-05 06:51:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5860b2b992772ba35e61f071418f28f18f5d9485","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-05 06:51:48.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"3a4c92194535ac7e8efbd0d79f1638f2f8e1c366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 06:52:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027gate\u0027][\u0027queue\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/infra-manual\n      check:\n        jobs:\n          - opendev-tox-docs\n      gate:\n        queue: testing\n        jobs:\n          - opendev-tox-docs\n      promote:\n        jobs:\n          - opendev-promote-docs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 104, column 3","accounts_in_message":[],"_revision_number":1},{"id":"56819b890a90b424356b7c39299af5615bcbe7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-05 07:01:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd2fe55282a6356e7c410d2a23de500b436418c6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-05 07:02:05.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            approval:\n              - Workflow: 1\n            require:\n              approval:\n                - Verified: [-1, -2]\n                  username: zuul\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"8568b0604aeb08398fbe8dffff580a6867af6abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 07:02:09.000000000","message":"Patch Set 2: Verified-1\n\n(10 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027gate\u0027][\u0027queue\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/infra-manual\n      check:\n        jobs:\n          - opendev-tox-docs\n      gate:\n        queue: testing\n        jobs:\n          - opendev-tox-docs\n      promote:\n        jobs:\n          - opendev-promote-docs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 104, column 3\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/project-config on branch master.  The problem\nwas:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following pipeline stanza:\n\n  pipeline:\n      name: release-approval\n      description: |\n        Newly-reviewed release requests enter this pipeline to check if the\n        current state can represent PTL or release liaison approval.\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: comment-added\n            username: ^(?!^zuul$)\\S+$\n      success:\n        gerrit:\n          PTL-Approved: 1\n          comment: false\n      failure:\n        gerrit:\n          PTL-Approved: 0\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 261, column 3","accounts_in_message":[],"_revision_number":2},{"id":"eb9cc771a0101ec80e7de5c74b6ce355fab98bc2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-05 07:04:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"10be3a3deb927546360f0e6688cfda7bea011837","revisions":{"704bc14ec85cb8136d34ce72a656072994be7106":{"kind":"REWORK","_number":1,"created":"2023-09-05 06:51:17.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/03/893703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/893703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/893703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/893703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/893703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/893703/1"}}},"commit":{"parents":[{"commit":"7ee4aab6c653a93c828e009a65d77abd6675495d","subject":"Update regexes in zuul config to the new style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ee4aab6c653a93c828e009a65d77abd6675495d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-05 06:50:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-05 06:50:59.000000000","tz":120},"subject":"DNM: testing zuul config error","message":"DNM: testing zuul config error\n\nChange-Id: I68ea25b380eda353f9e0ad0868bdf7f6730a6733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704bc14ec85cb8136d34ce72a656072994be7106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704bc14ec85cb8136d34ce72a656072994be7106"}]},"branch":"refs/heads/master"},"10be3a3deb927546360f0e6688cfda7bea011837":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-05 07:01:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/03/893703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/893703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/893703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/893703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/893703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/893703/2"}}},"commit":{"parents":[{"commit":"d278500d291cb6c10680810d1dfa11af374da849","subject":"Switch OpenStack\u0027s Zuul tenant to Ansible 8 by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d278500d291cb6c10680810d1dfa11af374da849"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-05 06:50:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-09-05 07:01:44.000000000","tz":0},"subject":"DNM: testing zuul config error","message":"DNM: testing zuul config error\n\nChange-Id: I68ea25b380eda353f9e0ad0868bdf7f6730a6733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10be3a3deb927546360f0e6688cfda7bea011837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10be3a3deb927546360f0e6688cfda7bea011837"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
