)]}'
{"id":"openstack%2Fproject-config~893912","triplet_id":"openstack%2Fproject-config~master~If11cc6ad901e1a45e96fb9b2b22ab4d8be01dfea","project":"openstack/project-config","branch":"master","topic":"modify-cinder-purestorage-acl","attention_set":{},"removed_from_attention_set":{"20870":{"account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"last_update":"2023-09-13 13:45:54.000000000","reason":"\u003cGERRIT_ACCOUNT_20870\u003e replied on the change","reason_account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}}},"hashtags":[],"change_id":"If11cc6ad901e1a45e96fb9b2b22ab4d8be01dfea","subject":"Switch charm-cinder-purestorage acl","status":"ABANDONED","created":"2023-09-06 13:16:33.000000000","updated":"2023-09-13 13:46:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"032b95dee5b51c0a1df52cfbe502b77925cd7b93","_number":893912,"virtual_id_number":893912,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-06 13:33:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 13:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 14:13:13.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"}],"messages":[{"id":"a31bb37c3c250a2fe07ce54022407294b90047dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-09-06 13:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39f3d2afc323017ae9368d1b8dab8404ae595276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-09-06 13:16:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"56addc7767215e64384cdd99ac55bd28cdb49d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 13:33:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4b974c1efc54615ba49bc7d695d6143\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7eeb0a7db91b407482a86194040977a9 : SUCCESS in 15m 38s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/540c70f3bcee49ef8f322ce007e6dd78 : SUCCESS in 3m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fff3132e462c49bc94ab9020d3c97a00 : SUCCESS in 6m 33s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/d526cde4d0154adb8406535dc353bb3f : SUCCESS in 4m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/01399a1238454bd5a60072a6cd706a74 : SUCCESS in 3m 55s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/f78bda7cf24c4b929e452b7347ea82f8 : SUCCESS in 12m 02s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/f1209c8ae88d49e3b391e605977d840c : SUCCESS in 3m 17s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/dfb2e46ca2394182bdab0d138d7c2111 : SUCCESS in 5m 47s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ff2d86243f5b4dbb97d22dabd21c4b28 : SUCCESS in 4m 34s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/85677c5625f945d69e76de9d19367141 : SUCCESS in 3m 36s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/6a284b8aeb88494eb05bda51eea4b400 : FAILURE in 2m 02s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a674d17eb6d446d18d958a61232d5a86 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f63ddbdc39c19acbad4600a5c058233741c339ab","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-06 14:13:13.000000000","message":"Patch Set 2:\n\nrecheck\n\nThe opensuse-15 test is failing with \"Repository \u0027repo-update\u0027 appears to be outdated. Consider using a different mirror or server. Package \u0027+libvirt-python\u0027 not found.\" That seems clearly unrelated to this change or its parent, and may represent a temporarily issue with OpenSuse package repositories or mirroring.","accounts_in_message":[],"_revision_number":2},{"id":"e034da650c891083b2625eff96fba300f34b8c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:42:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47843eba5b448d499175c2d7ba00980\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/120693aa82a54523b888b858394f30c2 : SUCCESS in 22m 34s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/fbf1c6a1a8c2494eab11ecd72b42dc35 : SUCCESS in 3m 32s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9cf47897719843a7ac190d578c245045 : SUCCESS in 7m 01s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/b1f25d3204ec4b76ab32b030979d5eae : SUCCESS in 3m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3109b7bc13d244a8bdc216fd34ac79aa : SUCCESS in 3m 59s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/42a567d860c34bae863b13ae51a6bf00 : SUCCESS in 12m 35s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/33549f8b47bd4dcdb0f5c07fed27a485 : SUCCESS in 4m 53s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/2e9287b74b8b4601961a58b67936cef6 : SUCCESS in 7m 21s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0ef4577968104552bbed391c566d6c2a : SUCCESS in 5m 21s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/79f12123375e4ea1bcf3c9dd11eaa669 : SUCCESS in 5m 20s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/cfe503710fa543618da6bf513c7576ba : FAILURE in 2m 47s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9b95bfcdfd6549f68ccf467b3b9ca605 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7b22c63be44528dd24a8186574aa5f7de742e384","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-06 15:14:21.000000000","message":"Patch Set 2:\n\nOh, also this change is empty. Was this leftover from a rebase maybe?","accounts_in_message":[],"_revision_number":2},{"id":"96fc3e87b3c67de866860cec85794139dda1bca8","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-09-13 13:45:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"032b95dee5b51c0a1df52cfbe502b77925cd7b93","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-09-13 13:46:42.000000000","message":"Abandoned\n\nalready done via I5465a232c10063f71e1d62f1d1d4d05b6ff64bf8","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff97b725c1b169e3e0627ef03613425fa5493503","revisions":{"a750467dd9178836fa5c7a37706619d2a9518691":{"kind":"REWORK","_number":1,"created":"2023-09-06 13:16:33.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/12/893912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/893912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/893912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/893912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/893912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/893912/1"}}},"commit":{"parents":[{"commit":"7ee4aab6c653a93c828e009a65d77abd6675495d","subject":"Update regexes in zuul config to the new style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ee4aab6c653a93c828e009a65d77abd6675495d"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2023-09-06 13:15:35.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2023-09-06 13:15:35.000000000","tz":60},"subject":"Switch charm-cinder-purestorage acl","message":"Switch charm-cinder-purestorage acl\n\nSwitch the charm-cinderr-purestorage acl from charm.config to\ncharm-purestorage.config to enable more control/collaboration with Pure\nStorage.\n\nChange-Id: If11cc6ad901e1a45e96fb9b2b22ab4d8be01dfea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a750467dd9178836fa5c7a37706619d2a9518691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a750467dd9178836fa5c7a37706619d2a9518691"}]},"branch":"refs/heads/master"},"ff97b725c1b169e3e0627ef03613425fa5493503":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-06 13:16:54.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/12/893912/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/893912/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/893912/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/893912/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/893912/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/893912/2"}}},"commit":{"parents":[{"commit":"7b604c060ba271f39fc7431b924ae15bbda65a01","subject":"Add charms-purestorage group for purestorage charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b604c060ba271f39fc7431b924ae15bbda65a01"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2023-09-06 13:15:35.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2023-09-06 13:16:54.000000000","tz":0},"subject":"Switch charm-cinder-purestorage acl","message":"Switch charm-cinder-purestorage acl\n\nSwitch the charm-cinderr-purestorage acl from charm.config to\ncharm-purestorage.config to enable more control/collaboration with Pure\nStorage.\n\nChange-Id: If11cc6ad901e1a45e96fb9b2b22ab4d8be01dfea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff97b725c1b169e3e0627ef03613425fa5493503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff97b725c1b169e3e0627ef03613425fa5493503"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
