)]}'
{"id":"openstack%2Fproject-config~893935","triplet_id":"openstack%2Fproject-config~master~I0bb7d9b7b34f4f6c392374182538b7e433617e13","project":"openstack/project-config","branch":"master","topic":"opensuse-libvirt","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-09-06 15:44:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0bb7d9b7b34f4f6c392374182538b7e433617e13","subject":"Drop libvirt-python from suse in bindep fallback","status":"MERGED","created":"2023-09-06 15:19:31.000000000","updated":"2023-09-06 15:51:06.000000000","submitted":"2023-09-06 15:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893935-opensuse-libvirt","meta_rev_id":"0439a4bbe82b772a6b79705b6e939e9d40ad7917","_number":893935,"virtual_id_number":893935,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 15:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-09-06 15:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-09-06 15:21:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 15:21:59.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-09-06 15:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"151cfd514658973f441e29abc78767d3a669ed3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-06 15:19:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f1a7d3badc031f03ce4960e9db3102e0daff15a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-09-06 15:21:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1471ab6d9cde324aba9cf30f8121ced5a6d92724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:32:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab7e439556e49e8abb4ed916a8659a3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d7649a5476d74f1d89606b6687297344 : SUCCESS in 8m 19s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/a941c0cbfdbf447b8dde8a7ed4f8d9fb : SUCCESS in 3m 16s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/63925b64c2a84a53a7b8613d9156177b : SUCCESS in 3m 03s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/7d293d9ba5e44acdb122e50a68b8caa2 : SUCCESS in 4m 08s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/06fa00def3164205a34f79a7cea8892a : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"49af09198bf768bb5ebe5cda173b29107142f36d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:34:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6ff60aa0ba118a5ccb2a8d36271a0d4e8521028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:44:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8360411dcd534ac5bbd3fde4b1f1b496\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3d65edd39b6a47c0b7e0d84f5af968b7 : SUCCESS in 6m 32s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/45534957b2f14a06962b91340101a20f : SUCCESS in 3m 19s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/8e542b8b752f4479ae193c6f50211270 : SUCCESS in 2m 53s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/ea0e58ae73974f13a4685b12936a551d : SUCCESS in 2m 51s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f66e2a4cee02464da0c2e83ec8f87fe1 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"27e6c9d8cb1f48ebc0dbc444075d8431df060389","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:44:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0439a4bbe82b772a6b79705b6e939e9d40ad7917","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a648afeac3cb429da2af263be31ec862\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/bf0407b63922448793aeea72f31fe8ce : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16ddb49e48de7e55a54a3cd1b819754560590d2d","revisions":{"16ddb49e48de7e55a54a3cd1b819754560590d2d":{"kind":"REWORK","_number":1,"created":"2023-09-06 15:19:31.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/35/893935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/893935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/893935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/893935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/893935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/893935/1"}}},"commit":{"parents":[{"commit":"76c8cd62ee5138214f6b3db0eefc3e94a07fd47e","subject":"Merge \"Add charms-purestorage group for purestorage charms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76c8cd62ee5138214f6b3db0eefc3e94a07fd47e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-09-06 15:08:51.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-09-06 15:15:03.000000000","tz":0},"subject":"Drop libvirt-python from suse in bindep fallback","message":"Drop libvirt-python from suse in bindep fallback\n\nThe bindep fallback list includes a libvirt-python package for all\nRPM-based distros, but it appears that OpenSuse Leap has recently\ndropped this (likely as part of removing Python 2.7 related\npackages). Exclude the package on that platform so that the\nopensuse-15 job will stop failing.\n\nChange-Id: I0bb7d9b7b34f4f6c392374182538b7e433617e13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16ddb49e48de7e55a54a3cd1b819754560590d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16ddb49e48de7e55a54a3cd1b819754560590d2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
