)]}'
{"id":"openstack%2Fproject-config~894289","triplet_id":"openstack%2Fproject-config~master~I2217158c0588071adecd25ed20c52e9fab023541","project":"openstack/project-config","branch":"master","topic":"charm-designate-k8s","attention_set":{},"removed_from_attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2023-09-19 13:15:05.000000000","reason":"Change was submitted"},"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2023-09-12 08:06:10.000000000","reason":"\u003cGERRIT_ACCOUNT_935\u003e replied on the change","reason_account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}}},"hashtags":[],"change_id":"I2217158c0588071adecd25ed20c52e9fab023541","subject":"Add OpenStack K8S Designate charms","status":"MERGED","created":"2023-09-08 09:35:20.000000000","updated":"2023-09-19 13:31:04.000000000","submitted":"2023-09-19 13:15:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894289-charm-designate-k8s","meta_rev_id":"84045dab2d6978d77abf0932db77a440344563ef","_number":894289,"virtual_id_number":894289,"owner":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 13:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2023-09-12 08:06:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-19 13:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-19 13:06:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 09:39:13.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2023-09-08 09:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-19 13:06:01.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"f68f476e8c25bd9682d8889c004ada0e4f4461cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-08 09:35:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f8c269e4777ffec629c3cae3c53f9a282729357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:42:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de4baf9568542408e5915be6e0b7743\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b4e24fa5683741ff8efe92c373c87e75 : SUCCESS in 6m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/82ae969c64114bbf8105604dddb44de8 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"81fd2e395039334c530e819e8fa51785cf7e440b","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-09-12 08:06:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c28651cc58ea3821b66b1926c6ca8ef95156187","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-19 13:06:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCorresponding governance change is https://review.opendev.org/894291 for reference.","accounts_in_message":[],"_revision_number":1},{"id":"bbe5f7d3dcd491ef951f455896b95e2be5a23388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 13:06:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ea944d4805eb7ecc21863ebf1c6c0ee83cd32d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 13:15:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef37d6caf044bc98c47d6d3ecd7b0e7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/615b44a468e44645ad7469b77e134ecc : SUCCESS in 5m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/684817d2fda74ec680212900d0039365 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5669233ec138b90d4b96c6cf6958bd3404c66118","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 13:15:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84045dab2d6978d77abf0932db77a440344563ef","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 13:31:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/759ad9beba8c4b56ab78d4b37a58626f\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/5cd770b6fc9e45b8bc5a879106fc89d7 : SUCCESS in 4m 23s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/92db13783d594f028096bc23b2e1faff : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa4f2781b439d9e28316700de5c6600e8e4b4df2","revisions":{"fa4f2781b439d9e28316700de5c6600e8e4b4df2":{"kind":"REWORK","_number":1,"created":"2023-09-08 09:35:20.000000000","uploader":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"ref":"refs/changes/89/894289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/89/894289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/894289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/894289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/89/894289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/89/894289/1"}}},"commit":{"parents":[{"commit":"8d32d45da269b44f6769440d098940e40b6c04e6","subject":"Set fedora labels min-ready to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d32d45da269b44f6769440d098940e40b6c04e6"}]}],"author":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-09-08 09:34:53.000000000","tz":120},"committer":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-09-08 09:34:53.000000000","tz":120},"subject":"Add OpenStack K8S Designate charms","message":"Add OpenStack K8S Designate charms\n\nSetup and import OpenStack K8S Designate Charm for deployment of\ndesignate hosted in Kubernetes.\n\nChange-Id: I2217158c0588071adecd25ed20c52e9fab023541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa4f2781b439d9e28316700de5c6600e8e4b4df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fa4f2781b439d9e28316700de5c6600e8e4b4df2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
