)]}'
{"id":"openstack%2Fproject-config~894419","triplet_id":"openstack%2Fproject-config~master~I5f1190718167817dc97f8d9126798c6471f0b78c","project":"openstack/project-config","branch":"master","topic":"retire-winstackers","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-09-18 17:24:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f1190718167817dc97f8d9126798c6471f0b78c","subject":"Retire os-win: remove project from infra","status":"MERGED","created":"2023-09-09 19:59:57.000000000","updated":"2023-09-18 17:43:28.000000000","submitted":"2023-09-18 17:24:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894419-retire-winstackers","meta_rev_id":"b537beaede7e73a1a15e6b6d41e7d02a344a0e40","_number":894419,"virtual_id_number":894419,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-18 17:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-10 17:52:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-18 17:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-09 20:00:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-09 20:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-10 17:52:37.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"9be53a896084f319fd76c7e40a8af1265694be69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-09 19:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a05f74284bcfa2ab30ca7f3ea68ba51fe52a43d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-09 20:00:22.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"f9cbaf0cb24e6d60df40301732b008601ab6967e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 20:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ae7d56ca0eb4aafaf98f8466917be04\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3ed3674078524ef38cca58c48083a7b1 : SUCCESS in 4m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b782c5b11c444618aeaeab91958d400d : SUCCESS in 3m 33s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e4ddb4c588fd47ac97a71491b1ff6955 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"079e8cee3f231f1eee27aa62b765cca25a01d2d6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-10 17:52:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ceb600b8e4809c4b35f894612931f0a05b11faa6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-18 16:57:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c028a588341a47985a77726f34686dcc04050d8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-18 17:12:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a1774bd87c1019efd26c988cb0c277518483a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51a2dac228a76f582d36fcf26f5f5610916eba03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:24:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c990d8fbc8e94bd7bd41075d8973398d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64243521611e4c1187f4384fd56b78e3 : SUCCESS in 6m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b98029995f934c1199c24d5be4336fa3 : SUCCESS in 3m 39s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a75f3b2f2a2a491594d007e4d70c2901 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e1375026bba14fcb5d61da3821f5471202b6d16e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:24:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b537beaede7e73a1a15e6b6d41e7d02a344a0e40","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:43:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981b913ad43745f7a30a0c80987686a4\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/3b9e85f7725847cda581608de326dab5 : SUCCESS in 4m 13s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/0550354755ca4b93a706d0c870b8df6e : SUCCESS in 9m 24s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/2ed84395022648fc9370399fb83cb502 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dffb8986253c50aa86b6fe16f9a55ee6b15e4cd6","revisions":{"dffb8986253c50aa86b6fe16f9a55ee6b15e4cd6":{"kind":"REWORK","_number":1,"created":"2023-09-09 19:59:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/894419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/894419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/894419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/894419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/894419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/894419/1"}}},"commit":{"parents":[{"commit":"299f5bf13262998149e9a799d83372feaece540c","subject":"Retire oswin-tempest-plugin: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/299f5bf13262998149e9a799d83372feaece540c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-09-09 19:58:32.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-09-09 19:58:32.000000000","tz":-420},"subject":"Retire os-win: remove project from infra","message":"Retire os-win: remove project from infra\n\nWinstackers project has been retired\n- https://review.opendev.org/c/openstack/governance/+/886880\n\nthis commit removes the os-win deliverables of this\nproject from infra.\n\nDepends-On: https://review.opendev.org/c/openstack/os-win/+/894411\nChange-Id: I5f1190718167817dc97f8d9126798c6471f0b78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dffb8986253c50aa86b6fe16f9a55ee6b15e4cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dffb8986253c50aa86b6fe16f9a55ee6b15e4cd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
