)]}'
{"id":"openstack%2Fproject-config~894605","triplet_id":"openstack%2Fproject-config~master~Iae482037cb417b121c61cf3bf8644eaac8cd541d","project":"openstack/project-config","branch":"master","topic":"add-manila-core","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-09-12 12:55:57.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-09-12 12:55:57.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-09-12 08:23:05.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-09-12 12:45:25.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iae482037cb417b121c61cf3bf8644eaac8cd541d","subject":"Add manila-core to osc/sdk repo config","status":"MERGED","created":"2023-09-11 23:21:38.000000000","updated":"2023-09-12 13:00:46.000000000","submitted":"2023-09-12 12:55:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894605-add-manila-core","meta_rev_id":"594102997e9e2cf0f1aa1b585a38eee5bdbc1172","_number":894605,"virtual_id_number":894605,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 12:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2023-09-12 08:23:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-12 12:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-12 12:45:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 23:22:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-09-11 23:22:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-09-11 23:22:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-09-11 23:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a585f4f5c21cad762a84db5c53b823a7f5fd5edf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-11 23:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ff440115e24740bc2b1efecf9e7eb9d6ec680f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 23:27:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c211375f9623408aa578b109417f71cc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/afe5b017248c4a6391c31047c1b735c5 : SUCCESS in 4m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3379d955722942578466f3ced5ac90a2 : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c385ad6a22d86bddd325611dc435dc0b3fc30758","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-11 23:34:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"638bd085f94871921213ee1a08ce7628fe0d45d7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 00:24:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01b26129a8067e091cc51fddc728755e40adbf27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 00:25:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2174051fb907785c5609944190d6989a51ec930c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-12 00:26:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4cda5d302e9f654dbb4f422350e02c4d52dde1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 00:40:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d37965b9284144b80fa6c9d1d5a717\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e4e1fdf9820045c3b66d7a3dfaa98611 : SUCCESS in 5m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8e126458159f46ccb4741ec21bfb8125 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"dc41fc35b745b47f9958d1e89a376c3a6acc6e11","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-09-12 08:23:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f7884a1646c857b62129b8f4d79548c7777b71c9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-12 12:45:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9d2bd5ab710b942a20457de2b83d40d0241f3e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 12:45:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f1f5c30af1a55d01e52d0b249b80dbb36dba4c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 12:55:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d123a5087446be85fdfe5df1b65660\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bc0b3bb53303416fb6d6df28b0f70b56 : SUCCESS in 8m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f260401dd1ed4c80ac07dd7ecf8665c7 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"eb52fcd689402fbb847f3082dc37b8308981741e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 12:55:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"594102997e9e2cf0f1aa1b585a38eee5bdbc1172","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:00:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d67c84a386248d9bfc85719aa7c17e7\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b3a28292c0544eb88867d7741b904073 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"232fb3df4cf78e3a403c3d02318c359bf7b3086c","revisions":{"857298ee2f4c32e2788056feae86fc670ea855ff":{"kind":"REWORK","_number":1,"created":"2023-09-11 23:21:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/05/894605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/894605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/894605/1"}}},"commit":{"parents":[{"commit":"aa48577b6a1ec864779864b603d1dbdea8d689af","subject":"Merge \"Remove fedora image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa48577b6a1ec864779864b603d1dbdea8d689af"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-11 23:20:42.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-11 23:20:42.000000000","tz":-420},"subject":"Add manila-core to osc/sdk repo config","message":"Add manila-core to osc/sdk repo config\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\nmanila-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Iae482037cb417b121c61cf3bf8644eaac8cd541d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/857298ee2f4c32e2788056feae86fc670ea855ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/857298ee2f4c32e2788056feae86fc670ea855ff"}]},"branch":"refs/heads/master"},"e37e466fdb3517b6b3f5701afeaae78ad3d934d5":{"kind":"REWORK","_number":2,"created":"2023-09-12 00:25:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/05/894605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/894605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/894605/2"}}},"commit":{"parents":[{"commit":"aa48577b6a1ec864779864b603d1dbdea8d689af","subject":"Merge \"Remove fedora image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa48577b6a1ec864779864b603d1dbdea8d689af"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-11 23:20:42.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-12 00:24:18.000000000","tz":-420},"subject":"Add manila-core to osc/sdk repo config","message":"Add manila-core to osc/sdk repo config\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\nmanila-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Iae482037cb417b121c61cf3bf8644eaac8cd541d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e37e466fdb3517b6b3f5701afeaae78ad3d934d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e37e466fdb3517b6b3f5701afeaae78ad3d934d5"}]},"branch":"refs/heads/master"},"232fb3df4cf78e3a403c3d02318c359bf7b3086c":{"kind":"REWORK","_number":3,"created":"2023-09-12 00:26:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/05/894605/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/894605/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/894605/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/894605/3"}}},"commit":{"parents":[{"commit":"aa48577b6a1ec864779864b603d1dbdea8d689af","subject":"Merge \"Remove fedora image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa48577b6a1ec864779864b603d1dbdea8d689af"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-11 23:20:42.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-12 00:26:12.000000000","tz":-420},"subject":"Add manila-core to osc/sdk repo config","message":"Add manila-core to osc/sdk repo config\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\nmanila-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Iae482037cb417b121c61cf3bf8644eaac8cd541d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/232fb3df4cf78e3a403c3d02318c359bf7b3086c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/232fb3df4cf78e3a403c3d02318c359bf7b3086c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
