)]}'
{"id":"openstack%2Fproject-config~894910","triplet_id":"openstack%2Fproject-config~master~I7eb00baed6646d77ec592414588272bdac43326a","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"24777":{"account":{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},"last_update":"2023-09-14 12:29:41.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2023-09-14 12:29:41.000000000","reason":"Change was submitted"},"28558":{"account":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"last_update":"2023-09-14 12:29:41.000000000","reason":"Change was submitted"},"35295":{"account":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"last_update":"2023-09-14 12:29:41.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-09-14 12:15:12.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7eb00baed6646d77ec592414588272bdac43326a","subject":"Add node-interface-metrics-exporter app to StarlingX","status":"MERGED","created":"2023-09-13 12:13:46.000000000","updated":"2023-09-14 12:40:24.000000000","submitted":"2023-09-14 12:29:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894910-new-project","meta_rev_id":"20a2c9b82ea00dd5658f4994001cbbf79a313737","_number":894910,"virtual_id_number":894910,"owner":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 12:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":2,"date":"2023-09-14 12:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":1,"date":"2023-09-14 12:15:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}],"CC":[{"_account_id":35510,"name":"Kailas Vitthalrao Surkar","email":"kailasvitthalrao.surkar@windriver.com","username":"kssurkar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 12:18:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-09-13 12:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-13 12:33:59.000000000","updated_by":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"reviewer":{"_account_id":35510,"name":"Kailas Vitthalrao Surkar","email":"kailasvitthalrao.surkar@windriver.com","username":"kssurkar"},"state":"CC"},{"updated":"2023-09-13 12:33:59.000000000","updated_by":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"CC"},{"updated":"2023-09-13 12:33:59.000000000","updated_by":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2023-09-13 12:56:09.000000000","updated_by":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"reviewer":{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},"state":"REVIEWER"},{"updated":"2023-09-13 13:24:51.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"REVIEWER"}],"messages":[{"id":"77dcab77fbc97edfe37ec18ffc34b19ec9d4f421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"date":"2023-09-13 12:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c495309f76647911281309f76a883a1a35d95d72","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-13 12:18:31.000000000","message":"Patch Set 1: Code-Review+2\n\nWe usually wait for Scott to acknowledge these with a +1 before approval.","accounts_in_message":[],"_revision_number":1},{"id":"d3cc01d22ac5ed9a59df6fc178c5c8601bbd71ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 12:25:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1d5fdb881945f1a20cd7d0bf7d8218\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6ef06f24f98e43b082e6ecc56130f9bd : SUCCESS in 6m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/84d68ef8f3aa4e15b9a38d829146038c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a7fecb005443e8f531ab223680c9320607991a9b","author":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"date":"2023-09-13 13:24:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e2f7fb3f3e1cf7d885e0222b1d764c2eef365833","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"date":"2023-09-13 13:25:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"07b078fdde83abfc5070f006f1009fccd52206e8","author":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"date":"2023-09-14 01:09:11.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19876fc732321ed27d61292aff62390540c5f9dd","author":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"date":"2023-09-14 01:09:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7708f80d2e85bc3a973df8aed7301182ffc7814b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"date":"2023-09-14 06:07:22.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b0bf2b34164ce60c2a944297409b71eff0c13df","author":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"date":"2023-09-14 06:08:13.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"b8cec29039f1055ca84f7ba8d240f63c8ced9390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 06:17:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a9ceec630c4a179c84b4ce2feaacab\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b14f92687fef4084aeab65fc72547226 : SUCCESS in 6m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/614edc859db74b0d94415f694d195575 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6cb9d1a52163e7c5aa1305519c93ac77b2752358","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-14 12:15:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThe additions requested by Ghada Khalil have been made, so I\u0027m taking that feedback as approval from StarlingX community leadership for adding this new repository to their namespace.","accounts_in_message":[],"_revision_number":2},{"id":"7ea171e0af82f6db1c9318483a83caaea843d674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:15:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a44f25157cc9b5318dcaec5ee8a1a231a744af52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:29:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c72e3d71024cc389ce86b743ea5001\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6eacac733c6442dab896521702c1887e : SUCCESS in 7m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/564bb49c8f044d1d8d76251389634404 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2f3073d2f16b421ca069c2fde778d1916bd6b3f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:29:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"20a2c9b82ea00dd5658f4994001cbbf79a313737","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:40:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d0f1a334ea46b0a82d6841471d7e89\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/8106215bda5b4b13a8937ed37599d8e1 : SUCCESS in 4m 06s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ba64466452b64964880bc2c709631eca : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad4b22554ecaefa5bd13962ce7e282316dd80290","revisions":{"36ac089ce88b898b50a95402164a59c50ed4a3ea":{"kind":"REWORK","_number":1,"created":"2023-09-13 12:13:46.000000000","uploader":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"ref":"refs/changes/10/894910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/894910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/894910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/894910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/894910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/894910/1"}}},"commit":{"parents":[{"commit":"50fe2484b1a8f857da94a327f83d8061184a7cfd","subject":"Merge \"Retire oswin-tempest-plugin: end project gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50fe2484b1a8f857da94a327f83d8061184a7cfd"}]}],"author":{"name":"Mayank Patel","email":"mayank.patel@windriver.com","date":"2023-09-13 12:12:24.000000000","tz":-240},"committer":{"name":"Mayank Patel","email":"mayank.patel@windriver.com","date":"2023-09-13 12:12:24.000000000","tz":-240},"subject":"Add node-interface-metrics-exporter app to StarlingX","message":"Add node-interface-metrics-exporter app to StarlingX\n\nChange-Id: I7eb00baed6646d77ec592414588272bdac43326a\nSigned-off-by: Mayank Patel \u003cmayank.patel@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36ac089ce88b898b50a95402164a59c50ed4a3ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36ac089ce88b898b50a95402164a59c50ed4a3ea"}]},"branch":"refs/heads/master"},"ad4b22554ecaefa5bd13962ce7e282316dd80290":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-14 06:07:22.000000000","uploader":{"_account_id":35295,"name":"Mayank Patel","display_name":"Mayank Patel","email":"mayank.patel@windriver.com","username":"mdpatel"},"ref":"refs/changes/10/894910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/894910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/894910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/894910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/894910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/894910/2"}}},"commit":{"parents":[{"commit":"50fe2484b1a8f857da94a327f83d8061184a7cfd","subject":"Merge \"Retire oswin-tempest-plugin: end project gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50fe2484b1a8f857da94a327f83d8061184a7cfd"}]}],"author":{"name":"Mayank Patel","email":"mayank.patel@windriver.com","date":"2023-09-13 12:12:24.000000000","tz":-240},"committer":{"name":"Mayank Patel","email":"mayank.patel@windriver.com","date":"2023-09-14 06:07:22.000000000","tz":0},"subject":"Add node-interface-metrics-exporter app to StarlingX","message":"Add node-interface-metrics-exporter app to StarlingX\n\nStory: 2010918\nTask: 48778\n\nChange-Id: I7eb00baed6646d77ec592414588272bdac43326a\nSigned-off-by: Mayank Patel \u003cmayank.patel@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad4b22554ecaefa5bd13962ce7e282316dd80290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad4b22554ecaefa5bd13962ce7e282316dd80290"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
