)]}'
{"id":"openstack%2Fproject-config~895809","triplet_id":"openstack%2Fproject-config~master~I5193a2ba497025a3ae1c09687dc953d30260991b","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-09-19 14:22:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5193a2ba497025a3ae1c09687dc953d30260991b","subject":"Add NebulOuS optimiser repos","status":"MERGED","created":"2023-09-19 11:51:54.000000000","updated":"2023-09-19 14:32:27.000000000","submitted":"2023-09-19 14:22:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895809-new-project","meta_rev_id":"569a4de673773e7b89e6881afbc3ca09ba3b5880","_number":895809,"virtual_id_number":895809,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 14:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-19 14:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-19 14:14:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-19 12:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-19 14:14:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"2ea923176873ce0446562e4525c69e3f687520c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-09-19 11:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b845aa8ea68b8a270984eb45ed5c425c9e3dd776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 12:03:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b87fc26129434eb6aa635329376163\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7ad2da368c474d358f69216c3c3805aa : SUCCESS in 6m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/016879ccbd1f4dafa51a9105892e3ca7 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"309a020017053da9274ef751aa7a2b1d136f9941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-09-19 13:34:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab1577f8aabf8ff077be28fc46e178f75252d9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 13:47:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32edd0456a3240ab99db155de04b8b84\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8712a41e880a4a5394b48c6277e365f7 : SUCCESS in 6m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/51d8378bb88b424cbbca7229a96cc06d : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e1785b981dc1e97908c85f59619677aad63f3efc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-19 14:14:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ff9dbe2ba0f5c37eb0789223acfcca461aff1c25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:15:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"199837f5a85fa2127cba0c620aa8fd0721cfc2bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:22:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024a0db40c114eb1b6d8bccfb8f37003\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa96be01fd8541f68b6dee3b79aa70a3 : SUCCESS in 6m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3a4ae80d76714c5cb10760b93d33b8e4 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"685643d36d278d9e3c2bc8922fbe79e3d0f72b35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:22:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"569a4de673773e7b89e6881afbc3ca09ba3b5880","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:32:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a82e6e4e2cf34398a94c2f5eeb5eedbe\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6d765ca8edd14d51a2109177a8bea307 : SUCCESS in 4m 21s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/c0dd547fee7d4424b4282027240b4690 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"427fb21c2197cfeba46ba23ae5f46a91ac2c0f5e","revisions":{"b426fca72845e97953965bd0b179c3bc678e77a6":{"kind":"REWORK","_number":1,"created":"2023-09-19 11:51:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/09/895809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/09/895809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/895809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/895809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/895809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/09/895809/1"}}},"commit":{"parents":[{"commit":"6bbe1ab0e5223f82c847306c354992e1c434bb72","subject":"Merge \"Retire oswin-tempest-plugin: remove project from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bbe1ab0e5223f82c847306c354992e1c434bb72"}]}],"author":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2023-09-19 11:34:55.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2023-09-19 11:49:43.000000000","tz":120},"subject":"Add NebulOuS optimiser repos","message":"Add NebulOuS optimiser repos\n\nChange-Id: I5193a2ba497025a3ae1c09687dc953d30260991b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b426fca72845e97953965bd0b179c3bc678e77a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b426fca72845e97953965bd0b179c3bc678e77a6"}]},"branch":"refs/heads/master"},"427fb21c2197cfeba46ba23ae5f46a91ac2c0f5e":{"kind":"REWORK","_number":2,"created":"2023-09-19 13:34:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/09/895809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/09/895809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/895809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/895809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/09/895809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/09/895809/2"}}},"commit":{"parents":[{"commit":"6bbe1ab0e5223f82c847306c354992e1c434bb72","subject":"Merge \"Retire oswin-tempest-plugin: remove project from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bbe1ab0e5223f82c847306c354992e1c434bb72"}]}],"author":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2023-09-19 11:34:55.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2023-09-19 13:34:04.000000000","tz":120},"subject":"Add NebulOuS optimiser repos","message":"Add NebulOuS optimiser repos\n\nChange-Id: I5193a2ba497025a3ae1c09687dc953d30260991b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/427fb21c2197cfeba46ba23ae5f46a91ac2c0f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/427fb21c2197cfeba46ba23ae5f46a91ac2c0f5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
