)]}'
{"id":"openstack%2Fproject-config~896070","triplet_id":"openstack%2Fproject-config~master~Id5b5364ef5675e86a6f14fdebc643ed4d8c75c69","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"35904":{"account":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"last_update":"2023-09-25 12:48:34.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-09-25 12:48:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5b5364ef5675e86a6f14fdebc643ed4d8c75c69","subject":"Add NebulOuS optimiser-utility-evaluator repo","status":"MERGED","created":"2023-09-25 08:45:55.000000000","updated":"2023-09-25 12:58:52.000000000","submitted":"2023-09-25 12:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896070-new-project","meta_rev_id":"0243a71867082f929a4a176ae1f1acca3f053746","_number":896070,"virtual_id_number":896070,"owner":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-25 12:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2023-09-25 09:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-25 12:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-25 12:42:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 08:46:45.000000000","updated_by":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2023-09-25 08:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 12:42:01.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"1f2c881b53ce059935242137b3769e93cf7afc0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2023-09-25 08:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16a4565ba816d32401447d4617d3e67f68c571be","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2023-09-25 08:46:40.000000000","message":"Topic set to new-project","accounts_in_message":[],"_revision_number":1},{"id":"0120ee3c182e67401b9956f2cc8f598580817cbf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-09-25 08:48:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b5ae924d3c04b37e0adbdaf8ee474953293b619","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-09-25 08:48:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfc21dd6c3422d2c57a73f9ad7bd25f3114d4f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 08:51:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48af2b42be0349c5b7eee07b61dd93f8\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6c1b7f14c97949578f4e942b90e5aaec : SUCCESS in 4m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5a30218ac0f44621a604c9d2fce2499a : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6ecdb117112216868f6bbe4ecd71d723480377be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2023-09-25 08:56:03.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1f5210b42f681bd562630910f97352867285314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2023-09-25 08:56:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52839d4e5d259bc134a0feef682ea58f2c1f2685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 09:04:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78dbeb42194a427da258fd703f8ac57d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/350aa22b435d46b4a15776e977d2b8fb : SUCCESS in 5m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d8b3391ac2704a0b840d40a9076419d7 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2968f7aa4a50465dd39063391f5b6a607fc1efca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-09-25 09:10:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ce369230113bf4d6aff75ca7885d415c0f5909b4","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2023-09-25 09:22:34.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ec896d1fde716418c7b824a5697970a19bea57db","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-25 12:42:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b2540b1c24c98575d4b8a4c62b6eb65118cfface","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:42:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b1dbf7a4ca67d0e72e9794807435b07dad3abdce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:48:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/179372abfedf4ba2aa40b723d13c35d5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c8b8f0dd26c4c90ac163a3798660b5b : SUCCESS in 5m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2ddc89269ba2403b80cddeb4e199d5de : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8bbb3e71056a91eb98b47be525fecb7d4c6aa721","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:48:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0243a71867082f929a4a176ae1f1acca3f053746","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:58:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9621059f4284444b4e5794d7166cd09\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/cc7383791a6a45a593ec874985035111 : SUCCESS in 4m 38s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/38671b811ca549e18b40be78f72e5030 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1803c8a3f804e0569d594e897452389b417aa11b","revisions":{"e1a265facc2007470614075942f99e1caf3c02da":{"kind":"REWORK","_number":1,"created":"2023-09-25 08:45:55.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/70/896070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/896070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/896070/1"}}},"commit":{"parents":[{"commit":"509218044a74958e164781cd7fd0c11fb89ca4c7","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509218044a74958e164781cd7fd0c11fb89ca4c7"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2023-09-25 08:43:39.000000000","tz":120},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2023-09-25 08:45:27.000000000","tz":120},"subject":"Add NebulOuS optimiser repo","message":"Add NebulOuS optimiser repo\n\nChange-Id: Id5b5364ef5675e86a6f14fdebc643ed4d8c75c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1a265facc2007470614075942f99e1caf3c02da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1a265facc2007470614075942f99e1caf3c02da"}]},"branch":"refs/heads/master"},"9257f69d9b6f3a5237329739859b5a3f95824628":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-25 08:56:03.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/70/896070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/896070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/896070/2"}}},"commit":{"parents":[{"commit":"509218044a74958e164781cd7fd0c11fb89ca4c7","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509218044a74958e164781cd7fd0c11fb89ca4c7"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2023-09-25 08:43:39.000000000","tz":120},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2023-09-25 08:54:49.000000000","tz":120},"subject":"Add NebulOuS optimiser-utility-evaluator repo","message":"Add NebulOuS optimiser-utility-evaluator repo\n\nChange-Id: Id5b5364ef5675e86a6f14fdebc643ed4d8c75c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9257f69d9b6f3a5237329739859b5a3f95824628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9257f69d9b6f3a5237329739859b5a3f95824628"}]},"branch":"refs/heads/master"},"1803c8a3f804e0569d594e897452389b417aa11b":{"kind":"REWORK","_number":3,"created":"2023-09-25 08:56:26.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/70/896070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/896070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/896070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/896070/3"}}},"commit":{"parents":[{"commit":"509218044a74958e164781cd7fd0c11fb89ca4c7","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509218044a74958e164781cd7fd0c11fb89ca4c7"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2023-09-25 08:43:39.000000000","tz":120},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2023-09-25 08:56:14.000000000","tz":120},"subject":"Add NebulOuS optimiser-utility-evaluator repo","message":"Add NebulOuS optimiser-utility-evaluator repo\n\nChange-Id: Id5b5364ef5675e86a6f14fdebc643ed4d8c75c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1803c8a3f804e0569d594e897452389b417aa11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1803c8a3f804e0569d594e897452389b417aa11b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
