)]}'
{"id":"openstack%2Fproject-config~896272","triplet_id":"openstack%2Fproject-config~master~I31fc59722b1f881e5031f51d218c39b4e5f5b92d","project":"openstack/project-config","branch":"master","topic":"charm-designate-bind-k8s","attention_set":{},"removed_from_attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2023-10-03 12:18:39.000000000","reason":"Change was submitted"},"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2023-10-03 08:16:53.000000000","reason":"\u003cGERRIT_ACCOUNT_935\u003e replied on the change","reason_account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}}},"hashtags":[],"change_id":"I31fc59722b1f881e5031f51d218c39b4e5f5b92d","subject":"Add Designate Bind K8S charm","status":"MERGED","created":"2023-09-22 15:55:37.000000000","updated":"2023-10-03 12:30:14.000000000","submitted":"2023-10-03 12:18:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896272-charm-designate-bind-k8s","meta_rev_id":"f2ade15399a0ef077a8fa42fc583d8443093b4bd","_number":896272,"virtual_id_number":896272,"owner":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-03 12:18:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2023-10-03 08:16:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-27 16:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-03 12:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 16:04:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 15:18:10.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2023-09-27 16:51:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"16b6e25f01e3aefb1964eee9e9fd6d09dc8892db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-22 15:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60e49f2c443a530cb97638926d728ba8d070ca8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:04:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5eaa33dac5c49d29a576a7033c9bd74\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ec9d3c3accb54249acef87b108cc310d : SUCCESS in 6m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0a3bfcfc70744b2f9007b35794b4df12 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2c7b9f9347f3ef39bed7ff53f6f328845c728fa4","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-25 11:58:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f073c68699ee601d8a260c63a07fda2062f1fb2","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-09-25 15:18:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"335a449d69418559a40dc2f43dbcb80c9e950b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-27 16:34:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a027aeb5fce8db620d35eb83d52f2bb4f4157869","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-27 16:34:27.000000000","message":"Topic charm-bind-k8s removed","accounts_in_message":[],"_revision_number":2},{"id":"5639d3272993ffa0ff8f5da14f67ffb191c98c63","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-27 16:34:33.000000000","message":"Topic set to charm-designate-bind-k8s","accounts_in_message":[],"_revision_number":2},{"id":"b68a8eee12f16cafdd1b31cf0169ae53f65e6845","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2023-09-27 16:36:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a95e863ba59570355404cf0bafd5726f425ac016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:46:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beff9ea7e9284103b0a6eabde1fde32c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ef7fba334cf429f8a8c9d0fb321d3aa : SUCCESS in 6m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e2e8c84782da47d1996f9b153c4b4d40 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7d732b278eeb93b494a51c79b1bd2a53d8f591d1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-27 16:51:31.000000000","message":"Patch Set 2: Code-Review+2\n\nWe can approve this once James confirms he\u0027s still comfortable with it following the name change.","accounts_in_message":[],"_revision_number":2},{"id":"6ef757762fe91911a3c654348716921f1286b51c","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-10-03 08:16:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5cf4031d6a0986d32fcadcbc5a482bc088596529","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-03 12:12:17.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"75684d358491f5ef077265332a205a5892cbca57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:12:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8951128a83cc4fccfea9d6448993060feb5a3d80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:18:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7320f80b14a5411086fc181cad1178a4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/90b33e09864b4c1595c524d174b240b1 : SUCCESS in 5m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1f3139965d764b7094dc896742aa8abe : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9497843908ec937d6e88fe28cea61d91d07ce453","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:18:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f2ade15399a0ef077a8fa42fc583d8443093b4bd","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:30:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/656acd5fdec14303b1f676dfe2608f29\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6e7f4a95660340bbb1151b30225a154c : SUCCESS in 4m 52s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/89fe8344d7d24f2d8ef4a37e978c508f : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46581a598a44f8f6d32a8275dc812a0a5f2aa5b0","revisions":{"a799aca833b7e15c188112874703817dd1d54c57":{"kind":"REWORK","_number":1,"created":"2023-09-22 15:55:37.000000000","uploader":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"ref":"refs/changes/72/896272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/896272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/896272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/896272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/896272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/896272/1"}}},"commit":{"parents":[{"commit":"509218044a74958e164781cd7fd0c11fb89ca4c7","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509218044a74958e164781cd7fd0c11fb89ca4c7"}]}],"author":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-09-22 15:54:02.000000000","tz":120},"committer":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-09-22 15:54:02.000000000","tz":120},"subject":"Add Bind K8S charm","message":"Add Bind K8S charm\n\nSetup and import Bind K8S Charm for deployment of\nbind hosted in Kubernetes.\n\nBind is the dns backend of charm-designate-k8s.\n\nChange-Id: I31fc59722b1f881e5031f51d218c39b4e5f5b92d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a799aca833b7e15c188112874703817dd1d54c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a799aca833b7e15c188112874703817dd1d54c57"}]},"branch":"refs/heads/master"},"46581a598a44f8f6d32a8275dc812a0a5f2aa5b0":{"kind":"REWORK","_number":2,"created":"2023-09-27 16:34:20.000000000","uploader":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"ref":"refs/changes/72/896272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/72/896272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/896272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/896272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/72/896272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/72/896272/2"}}},"commit":{"parents":[{"commit":"509218044a74958e164781cd7fd0c11fb89ca4c7","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509218044a74958e164781cd7fd0c11fb89ca4c7"}]}],"author":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-09-22 15:54:02.000000000","tz":120},"committer":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2023-09-27 16:34:01.000000000","tz":120},"subject":"Add Designate Bind K8S charm","message":"Add Designate Bind K8S charm\n\nSetup and import Designate Bind K8S Charm for deployment of\nbind hosted in Kubernetes.\n\nBind is the dns backend of charm-designate-k8s.\n\nChange-Id: I31fc59722b1f881e5031f51d218c39b4e5f5b92d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46581a598a44f8f6d32a8275dc812a0a5f2aa5b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46581a598a44f8f6d32a8275dc812a0a5f2aa5b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
