)]}'
{"id":"openstack%2Fproject-config~897273","triplet_id":"openstack%2Fproject-config~master~Idfba1e793a6120df40c23da5ddc0dcc091e5d63e","project":"openstack/project-config","branch":"master","topic":"project-yaml-normalization","hashtags":[],"change_id":"Idfba1e793a6120df40c23da5ddc0dcc091e5d63e","subject":"Normalize projects.yaml","status":"MERGED","created":"2023-10-04 02:14:40.000000000","updated":"2023-10-06 03:09:22.000000000","submitted":"2023-10-06 03:00:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897273-project-yaml-normalization","meta_rev_id":"238b95577269e2fe6c002eedf18cf6573a5b4669","_number":897273,"virtual_id_number":897273,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 03:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-10-06 02:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-06 02:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-06 02:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 02:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 02:39:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-10-06 02:36:02.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"870cf446bcfe5e4e0227dc35f3f058456ff2cbfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-10-04 02:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af836f9699833981aaf11100a72ba6ff0aef20e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 02:40:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce23d13994c244a9bea2b675d5ab8637\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e8b31df2b51848d9bbf79fe769cbe602 : SUCCESS in 7m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/91eff1b128014833a67086ba1a5356fc : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2ce9e5e46614b831e3e8c6fec5e64867b14428f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-10-05 02:14:47.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4fb181dd7e05c9d82755f3fd51fa4d2eb3f7b176","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-05 02:39:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eaa4615a90c257576af296b4c9256210f6e046ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 02:48:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56fffc81b412473386b5ec1eac660d4c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/627555cd68234424ac730a502b614b3c : SUCCESS in 6m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c69465149bac4701bf294b97b36d8b4e : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4f7da5dde44cd52c3c8d67a6b7c8e58d262542f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-10-06 02:18:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d61950af1ec7c2915a7422b651dfe63620743048","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-06 02:36:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThe script which generates these doesn\u0027t retrieve the prior patchset like other proposal jobs. It could be augmented to do so if anyone finds time.","accounts_in_message":[],"_revision_number":3},{"id":"7bb6f4f101dd4a47ac0c1c03ecb68c40ece79b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 02:48:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/004cbae968934282a80a81818fda8179\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4c800d7d84a94bb3b8f4ac26c682ab01 : SUCCESS in 8m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bd76a34204224cc1bf1eab1e2cc0dc25 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"007013aec0271f8406931226454a3ca6b43bf357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 02:48:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"93547e3adf703ccd5a53947d0a9a135ca654a49c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 03:00:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962d97d512a6468d804741a9d918b7e9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/138dba3035e1422f8d0ec734be02bfa8 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/532320cb98474e4aac44d573e4b10117 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"22835eb63802f4613d9533863171ec54a6cac5a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 03:00:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"238b95577269e2fe6c002eedf18cf6573a5b4669","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 03:09:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4819c64065542eab4111efdcea1e7be\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6491f81cb6c24f44952657b6d557fd9e : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c847cfa2831154668a1586cdd57dc0fd57019ef","revisions":{"f3b7bf31e20014902a6c734aab103f027954d0e7":{"kind":"REWORK","_number":1,"created":"2023-10-04 02:14:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/897273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/897273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/897273/1"}}},"commit":{"parents":[{"commit":"d9bf2c6f4d4744a6975b9582997767ab3fa9a59a","subject":"Merge \"Temporarily remove release docs semaphores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9bf2c6f4d4744a6975b9582997767ab3fa9a59a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-04 02:14:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-04 02:14:37.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: Idfba1e793a6120df40c23da5ddc0dcc091e5d63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3b7bf31e20014902a6c734aab103f027954d0e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3b7bf31e20014902a6c734aab103f027954d0e7"}]},"branch":"refs/heads/master"},"54cd8a1f5dea64334a0499507459b25c0de72f05":{"kind":"NO_CHANGE","_number":2,"created":"2023-10-05 02:14:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/897273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/897273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/897273/2"}}},"commit":{"parents":[{"commit":"d9bf2c6f4d4744a6975b9582997767ab3fa9a59a","subject":"Merge \"Temporarily remove release docs semaphores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9bf2c6f4d4744a6975b9582997767ab3fa9a59a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-05 02:14:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-05 02:14:45.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: Idfba1e793a6120df40c23da5ddc0dcc091e5d63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/54cd8a1f5dea64334a0499507459b25c0de72f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/54cd8a1f5dea64334a0499507459b25c0de72f05"}]},"branch":"refs/heads/master"},"2c847cfa2831154668a1586cdd57dc0fd57019ef":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-06 02:18:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/897273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/897273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/897273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/897273/3"}}},"commit":{"parents":[{"commit":"201b50ef7639cf82db56b92b640e83e531aa3f50","subject":"Merge \"Clean up vestigial default-branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/201b50ef7639cf82db56b92b640e83e531aa3f50"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-06 02:17:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-10-06 02:17:59.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: Idfba1e793a6120df40c23da5ddc0dcc091e5d63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c847cfa2831154668a1586cdd57dc0fd57019ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c847cfa2831154668a1586cdd57dc0fd57019ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
