)]}'
{"id":"openstack%2Fproject-config~898118","triplet_id":"openstack%2Fproject-config~master~I8b87ab6d52e0ec7da92f80a62c7d97e4f2245b29","project":"openstack/project-config","branch":"master","topic":"projects-linter","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-10-12 14:10:21.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-11-28 23:07:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b87ab6d52e0ec7da92f80a62c7d97e4f2245b29","subject":"Check proposed project names for problems","status":"MERGED","created":"2023-10-12 13:36:22.000000000","updated":"2023-11-28 23:07:56.000000000","submitted":"2023-11-28 23:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"898118-projects-linter","meta_rev_id":"849005e0e6037dfb961afaec5e3b85873264c114","_number":898118,"virtual_id_number":898118,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 23:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-11-28 23:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2023-11-08 07:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-10-12 14:10:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-11-28 23:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 13:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 13:56:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2023-11-08 07:34:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-11-28 23:00:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"a94c26ef04c530fa6591b593ac17c584a751e6e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-12 13:36:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83ef6e56782c2c02f684b0c2a22d5af25277e798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 13:46:45.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bce1c2116534e5c92e3a5675c124703\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1ff2ffb92f304586963d80f8c0827f6d : FAILURE in 6m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/320dd88f6e2c4354b2749d6493fb05f9 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"eb199742f9cce8aa9b70402bc550c5858941ba53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-10-12 13:56:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"af1f6070e2a1acaebe31523bca5de5eabf56bab5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-10-12 13:56:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"df9454fdc25774fba309a7873fce00ca672ea5b1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-10-12 13:58:25.000000000","message":"Patch Set 2: -Code-Review\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a029bfdd4a892e4ff314e3a5afd77a36abec1d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 14:03:51.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3749c75ab5841aaa7cd2df39d9410a6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/43fa10cbfe9e42578e20c571d826d83e : FAILURE in 5m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/454978585a724eff909ad2ddece79c3a : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"088c7cc2e27cccb32aea15190c924dad956faecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-12 14:08:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1fe30c02e62a2b4b720e66e8265d5697b641159f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-10-12 14:10:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b947db99ac943a4075e130bdc76469424c1f7f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 14:19:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570122d5a0e44079a63a1c4f66b700f4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b7d0c5a82454705877e85155cb4bc4e : SUCCESS in 6m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a5eb75a25646453da75ac19f14af70aa : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"7790b87e6f7f339b745f88f3b446f2098e684288","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-08 07:34:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6b9cedf229e15e0902a1b6851665a563848b43ac","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-28 23:00:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"182f3e08a4fb3376582c483ce95f5c6ab0cc349b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 23:00:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b770b81f74c128e6adbeafac526136de42583176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 23:07:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ab2043d25142178440d7c2dd1699a5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/25739d5e8e544d848cf72690ebc452a2 : SUCCESS in 6m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6a14756795b0415ba81c7ec425095f04 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"849005e0e6037dfb961afaec5e3b85873264c114","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 23:07:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1c0f48cae156e329a16e7c68c0f146fba7880baa","revisions":{"10102171fc73a283a51910d98a25f7fc3387af54":{"kind":"REWORK","_number":1,"created":"2023-10-12 13:36:22.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/18/898118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/898118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/898118/1"}}},"commit":{"parents":[{"commit":"807eabdf4203bdb95645fa09433716275d4cb280","subject":"Add NebulOuS repos: exn and gui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/807eabdf4203bdb95645fa09433716275d4cb280"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-10-12 13:34:14.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-10-12 13:34:14.000000000","tz":0},"subject":"Check proposed project names for problems","message":"Check proposed project names for problems\n\nToday we discovered that Gitea doesn\u0027t allow creating projects with\n\"+\" in their names (or that it needs special escaping in some way).\nIn order to avoid future incidents, let\u0027s be defensive and limit\norganization/project names to only the characters which are\ncurrently in use, since we know those work. If people want to use\ncharacters outside that list in the future, we can test for\npotential breakage before amending the filter.\n\nChange-Id: I8b87ab6d52e0ec7da92f80a62c7d97e4f2245b29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10102171fc73a283a51910d98a25f7fc3387af54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10102171fc73a283a51910d98a25f7fc3387af54"}]},"branch":"refs/heads/master"},"626ecd1887c4c0547c1200a9a0614ba16047122c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-12 13:56:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/18/898118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/898118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/898118/2"}}},"commit":{"parents":[{"commit":"80711d4d417916e7729b9b32e3f19cdb62ebdeb8","subject":"Fix c++ to cpp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/80711d4d417916e7729b9b32e3f19cdb62ebdeb8"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-10-12 13:34:14.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2023-10-12 13:56:23.000000000","tz":0},"subject":"Check proposed project names for problems","message":"Check proposed project names for problems\n\nToday we discovered that Gitea doesn\u0027t allow creating projects with\n\"+\" in their names (or that it needs special escaping in some way).\nIn order to avoid future incidents, let\u0027s be defensive and limit\norganization/project names to only the characters which are\ncurrently in use, since we know those work. If people want to use\ncharacters outside that list in the future, we can test for\npotential breakage before amending the filter.\n\nChange-Id: I8b87ab6d52e0ec7da92f80a62c7d97e4f2245b29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/626ecd1887c4c0547c1200a9a0614ba16047122c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/626ecd1887c4c0547c1200a9a0614ba16047122c"}]},"branch":"refs/heads/master","description":"Rebase"},"1c0f48cae156e329a16e7c68c0f146fba7880baa":{"kind":"REWORK","_number":3,"created":"2023-10-12 14:08:22.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/18/898118/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/898118/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/898118/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/898118/3"}}},"commit":{"parents":[{"commit":"80711d4d417916e7729b9b32e3f19cdb62ebdeb8","subject":"Fix c++ to cpp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/80711d4d417916e7729b9b32e3f19cdb62ebdeb8"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-10-12 13:34:14.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-10-12 14:08:11.000000000","tz":0},"subject":"Check proposed project names for problems","message":"Check proposed project names for problems\n\nToday we discovered that Gitea doesn\u0027t allow creating projects with\n\"+\" in their names (or that it needs special escaping in some way).\nIn order to avoid future incidents, let\u0027s be defensive and limit\norganization/project names to only the characters which are\ncurrently in use, since we know those work. If people want to use\ncharacters outside that list in the future, we can test for\npotential breakage before amending the filter.\n\nChange-Id: I8b87ab6d52e0ec7da92f80a62c7d97e4f2245b29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c0f48cae156e329a16e7c68c0f146fba7880baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c0f48cae156e329a16e7c68c0f146fba7880baa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
