)]}'
{"id":"openstack%2Fproject-config~898711","triplet_id":"openstack%2Fproject-config~master~I1ead3f9f5944b6fe41e40fa63d25b50eebb0e777","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"35857":{"account":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"last_update":"2023-10-20 06:44:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1ead3f9f5944b6fe41e40fa63d25b50eebb0e777","subject":"Add Deployment-Manager app to StarlingX","status":"ABANDONED","created":"2023-10-18 07:03:25.000000000","updated":"2023-10-20 06:44:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ebdbf2ff3ee3e1668f5fe7be5e17dbbc50a8c2e","_number":898711,"virtual_id_number":898711,"owner":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-18 07:56:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-18 07:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54b6bea4d0b1b45702a2c7df394cc82c7cc88029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"date":"2023-10-18 07:03:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3afba44122f9a182ff56ad1e2155ee0306841dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:10:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/160d03409d1d49df99c44cc91843bb68\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/317e3d885d824d608673ed358bfb3c2c : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d388da20e075475b88dbfe0dee13e5a2 : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2e2e05027a0a569747a37cd4540840b353e3415c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"date":"2023-10-18 07:49:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f9e462cefa4f833623612acc48935845617e0c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:56:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0580e666982b45679cf85896e7df108f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ad814e37159741c28dbf880c498c7fe4 : SUCCESS in 6m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/553d073164d841ccad818debd8249b10 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8ebdbf2ff3ee3e1668f5fe7be5e17dbbc50a8c2e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"date":"2023-10-20 06:44:20.000000000","message":"Abandoned\n\nThis repo not be created in opendev.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fb57919de2b39d49ae6b5e29ef0c265895e030c","revisions":{"56c12d18e5047f9780431658d7a502d593b0052f":{"kind":"REWORK","_number":1,"created":"2023-10-18 07:03:25.000000000","uploader":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"ref":"refs/changes/11/898711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/898711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/898711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/898711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/898711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/898711/1"}}},"commit":{"parents":[{"commit":"4b41f9c7ef3891e9c6c23966aba90617656cd8bd","subject":"Merge \"Retire opendev/puppet-openstack_health: remove project from infa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b41f9c7ef3891e9c6c23966aba90617656cd8bd"}]}],"author":{"name":"Sharang Bhardwaj","email":"sharang.bhardwaj@windriver.com","date":"2023-10-18 07:01:45.000000000","tz":-240},"committer":{"name":"Sharang Bhardwaj","email":"sharang.bhardwaj@windriver.com","date":"2023-10-18 07:03:24.000000000","tz":-240},"subject":"Add Deployment-Manager app to StarlingX","message":"Add Deployment-Manager app to StarlingX\n\nChange-Id: I1ead3f9f5944b6fe41e40fa63d25b50eebb0e777\nSigned-off-by: Sharang Bhardwaj \u003csharang.bhardwaj@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56c12d18e5047f9780431658d7a502d593b0052f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56c12d18e5047f9780431658d7a502d593b0052f"}]},"branch":"refs/heads/master"},"6fb57919de2b39d49ae6b5e29ef0c265895e030c":{"kind":"REWORK","_number":2,"created":"2023-10-18 07:49:09.000000000","uploader":{"_account_id":35857,"name":"Sharang Bhardwaj","email":"Sharang.Bhardwaj@windriver.com","username":"sbhardwa"},"ref":"refs/changes/11/898711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/898711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/898711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/898711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/898711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/898711/2"}}},"commit":{"parents":[{"commit":"4b41f9c7ef3891e9c6c23966aba90617656cd8bd","subject":"Merge \"Retire opendev/puppet-openstack_health: remove project from infa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b41f9c7ef3891e9c6c23966aba90617656cd8bd"}]}],"author":{"name":"Sharang Bhardwaj","email":"sharang.bhardwaj@windriver.com","date":"2023-10-18 07:01:45.000000000","tz":-240},"committer":{"name":"Sharang Bhardwaj","email":"sharang.bhardwaj@windriver.com","date":"2023-10-18 07:44:23.000000000","tz":-240},"subject":"Add Deployment-Manager app to StarlingX","message":"Add Deployment-Manager app to StarlingX\n\nChange-Id: I1ead3f9f5944b6fe41e40fa63d25b50eebb0e777\nSigned-off-by: Sharang Bhardwaj \u003csharang.bhardwaj@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6fb57919de2b39d49ae6b5e29ef0c265895e030c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6fb57919de2b39d49ae6b5e29ef0c265895e030c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
