)]}'
{"id":"openstack%2Fproject-config~898836","triplet_id":"openstack%2Fproject-config~master~Ic7e9739fd016f00139cf617e67196b03eaef6a12","project":"openstack/project-config","branch":"master","topic":"neutron-dashboard","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-27 06:29:53.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-31 09:32:18.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-31 09:32:18.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-27 14:39:55.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ic7e9739fd016f00139cf617e67196b03eaef6a12","subject":"Update Grafana dashboard for Neutron master","status":"MERGED","created":"2023-10-19 15:32:23.000000000","updated":"2023-10-31 09:34:31.000000000","submitted":"2023-10-31 09:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898836-neutron-dashboard","meta_rev_id":"08f0fa353d6c7f23c379b4a5e96afba5571d98ee","_number":898836,"virtual_id_number":898836,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 09:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-10-27 06:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-10-27 14:39:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-10-31 09:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-19 15:43:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-10-19 15:43:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-19 15:43:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-19 15:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 13:37:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"824957d24ca7c099750442d34cc921fbf5a40d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-19 15:32:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d635a88931a0ce124a0188d1cced6c08b5d92b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 15:58:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e70251e59f8746ae918f673854adf2dc\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/03aa9b25eb5d46d6a46d860593cb202f : SUCCESS in 20m 49s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/343686e22fd84bf1a765764b6cd3c293 : SUCCESS in 6m 35s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/e076502843f8402fa6c24e88bec5a2bd : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5525dcf75fa10ce1b280fd65fa255b86e4035725","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-19 18:34:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"323b0565b479b95f848d2f411e21f797e248f5f6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-23 16:50:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f7419380beda9ab7ac3243726cca6725d438238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-26 08:25:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"72c8596e9e371f31dee868b4566d7cd1685c92d5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-26 08:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f47c9cfd9ab0f8ed6c0b4234b724469ad5c3774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 08:47:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf7f2aa854545f2b366ace3b99f6320\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a54f67b53c644caa920165e979759587 : SUCCESS in 19m 46s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eed8d717e90a46c2af88724db1527998 : SUCCESS in 6m 46s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/134ce7c912334d788f325ff26687c3e1 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4accfa13b981abcae963476f35b4264e93dcdc81","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-26 13:37:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b519af0a97f17ed883633e558c662b55edb95eb8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-26 18:25:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"823f9a57d5a4b3762548008359213507536a81dc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-26 18:28:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2fc09afa13659b604236550566deebe8425c78c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-26 18:47:51.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"8f779feba8165d82efe0dd686ee6c1c8574121e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-27 04:49:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"381680d146b76dc9ff65c2ebe703b12f3f938e77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-27 04:51:15.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d617005001155b2d3b1e9d2be2ac090aed9cacbe","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-27 04:51:34.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b9e3caca9d7c818afcfa1df06ad953a0885c7650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 05:08:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5a00c2d7674434833293b15ab18b74\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8d60fcf0777c4feb99f2cc65abf4882a : SUCCESS in 16m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2eda787623a04d5298cbaec51653170f : SUCCESS in 5m 08s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/6779affbe67944dea9baaa72277cc69f : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":4},{"id":"91afda66b31b3f849f06e599e10a003b7e7a2c6e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-27 06:29:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"302b7134f369c1ce43bdb8a15319df0dd05b57c9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 14:39:55.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"515b303dc66cd47a17c16dfa6a8a78fb260f789b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-31 09:06:00.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5ca00d01d5a0078158924ea4ea6a18e6e97c3a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:06:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8ff86b4389c532c1e7f1ff4091c1f41e4ed344ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:32:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13cc8e7b8bd643acb99bc8d6152d6ef0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a2c8d60d88264e99b1a4a0acd1c0496e : SUCCESS in 23m 10s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63182e3b80984f0991b091cdb41e61e5 : SUCCESS in 7m 25s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/1984daf7a54f431b89b06a201c30f04b : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2c00e622bdcb9ca288af8ff01464e4d34a845efb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:32:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"08f0fa353d6c7f23c379b4a5e96afba5571d98ee","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:34:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d14890acf7e404faebf559d30f5f946\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/ca61784d2c634a608661ebbb98c34cf6 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"45f0c8c4c90454a836ccc86de0984742acacb4a3","revisions":{"8d117d94ee0e6500be534f9db18ee3b0eae7063c":{"kind":"REWORK","_number":1,"created":"2023-10-19 15:32:23.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/36/898836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/898836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/898836/1"}}},"commit":{"parents":[{"commit":"4b41f9c7ef3891e9c6c23966aba90617656cd8bd","subject":"Merge \"Retire opendev/puppet-openstack_health: remove project from infa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b41f9c7ef3891e9c6c23966aba90617656cd8bd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-19 15:29:04.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-19 15:31:18.000000000","tz":330},"subject":"Update Grafana dasboard for Neutron master","message":"Update Grafana dasboard for Neutron master\n\nUpdate dashboard as per recent changes in the neutron\njob pipelines.\n\nChange-Id: Ic7e9739fd016f00139cf617e67196b03eaef6a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d117d94ee0e6500be534f9db18ee3b0eae7063c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d117d94ee0e6500be534f9db18ee3b0eae7063c"}]},"branch":"refs/heads/master"},"9307070aefee8a43fe7b89395d7e99b21c1097fe":{"kind":"REWORK","_number":2,"created":"2023-10-26 08:25:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/36/898836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/898836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/898836/2"}}},"commit":{"parents":[{"commit":"4b41f9c7ef3891e9c6c23966aba90617656cd8bd","subject":"Merge \"Retire opendev/puppet-openstack_health: remove project from infa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b41f9c7ef3891e9c6c23966aba90617656cd8bd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-19 15:29:04.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-26 08:24:57.000000000","tz":330},"subject":"Update Grafana dasboard for Neutron master","message":"Update Grafana dasboard for Neutron master\n\nUpdate dashboard as per recent changes in the neutron\njob pipelines.\n\nChange-Id: Ic7e9739fd016f00139cf617e67196b03eaef6a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9307070aefee8a43fe7b89395d7e99b21c1097fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9307070aefee8a43fe7b89395d7e99b21c1097fe"}]},"branch":"refs/heads/master"},"b2830f3c004700496069d98d5b280fea01047c87":{"kind":"REWORK","_number":3,"created":"2023-10-27 04:49:45.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/36/898836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/898836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/898836/3"}}},"commit":{"parents":[{"commit":"4b41f9c7ef3891e9c6c23966aba90617656cd8bd","subject":"Merge \"Retire opendev/puppet-openstack_health: remove project from infa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b41f9c7ef3891e9c6c23966aba90617656cd8bd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-19 15:29:04.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-27 04:48:41.000000000","tz":330},"subject":"Update Grafana dasboard for Neutron master","message":"Update Grafana dasboard for Neutron master\n\nUpdate dashboard as per recent changes in the neutron\njob pipelines.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/899450\nChange-Id: Ic7e9739fd016f00139cf617e67196b03eaef6a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2830f3c004700496069d98d5b280fea01047c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2830f3c004700496069d98d5b280fea01047c87"}]},"branch":"refs/heads/master"},"45f0c8c4c90454a836ccc86de0984742acacb4a3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-10-27 04:51:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/36/898836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/898836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/898836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/898836/4"}}},"commit":{"parents":[{"commit":"4b41f9c7ef3891e9c6c23966aba90617656cd8bd","subject":"Merge \"Retire opendev/puppet-openstack_health: remove project from infa\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b41f9c7ef3891e9c6c23966aba90617656cd8bd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-19 15:29:04.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-27 04:50:57.000000000","tz":330},"subject":"Update Grafana dashboard for Neutron master","message":"Update Grafana dashboard for Neutron master\n\nUpdate dashboard as per recent changes in the neutron\njob pipelines.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/899450\nChange-Id: Ic7e9739fd016f00139cf617e67196b03eaef6a12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45f0c8c4c90454a836ccc86de0984742acacb4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45f0c8c4c90454a836ccc86de0984742acacb4a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
