)]}'
{"id":"openstack%2Fproject-config~900167","triplet_id":"openstack%2Fproject-config~master~I79e5efca77a49b1cd2b0706d37a39cbbab4a19dd","project":"openstack/project-config","branch":"master","topic":"sunbeam-charm-mono-repo","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-10 16:23:10.000000000","reason":"Change was submitted"},"12549":{"account":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"last_update":"2023-11-10 16:23:10.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-11-06 14:49:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I79e5efca77a49b1cd2b0706d37a39cbbab4a19dd","subject":"Create mono repo for sunbeam charms","status":"MERGED","created":"2023-11-06 11:29:46.000000000","updated":"2023-11-10 16:38:14.000000000","submitted":"2023-11-10 16:23:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900167-sunbeam-charm-mono-repo","meta_rev_id":"324e47ce23b89cf7016db214d44fce4d1f848902","_number":900167,"virtual_id_number":900167,"owner":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 16:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2023-11-07 06:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-06 14:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-10 16:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}],"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 11:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 14:24:15.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-11-06 14:39:34.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-11-06 14:42:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-07 06:14:45.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2023-11-07 07:53:30.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"state":"CC"}],"messages":[{"id":"fb1b7f2525acfdce065c2390075b602b11aa68dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 11:29:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2290a31a4d57e543a8ade7362dc47b2be740c39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:44:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a23e27a707ae422ea422cf16dfdb7238\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b57d0f55e0fe4d1ebf0bfef30db39501 : SUCCESS in 6m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/73cf2cae87d9492dacdefb23799e53b3 : FAILURE in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f74776c6b43f3ea2e59db2be04a8974e27b9111c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 12:16:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"697122667015a81078ff1bfd5878c4fe70ee19d1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 13:51:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"7d892c7edd0e1333f774162d329567ceca32b88b","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 13:51:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3d4780881fcc1a0bd3fd1ef352811f383c7ff9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af7e679fa764a6eb4c747219c1b7651\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/924f8c8a5e604a0b8229fd964097f8c5 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a7d61da91dea4fdf93fc6d77d07d4b31 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"27850414e4bc5ff3e1ef8787d870f1652e001a55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 14:04:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c5943c11c93ee8957209719c92e98793696b849c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:13:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a049e115f1c4aafbddb0f8340923012\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bd36e57d68ae4b99993471f36ec24270 : SUCCESS in 7m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c22b15b4202f40e6a86592346e69171e : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"3d3371a7a46a9a860fdf9c0735dbf4a83eba4173","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-06 14:24:15.000000000","message":"Patch Set 2:\n\nJust so I\u0027m clear on the plan, you\u0027re going to graft the Git histories of the existing individual repos together so it isn\u0027t lost, then once this is created you\u0027re going to close any open changes for the individual repos and retire each one? You\u0027ll also need to pick some point in time to freeze activity on the existing repos so that your merged version of them doesn\u0027t get out of sync.","accounts_in_message":[],"_revision_number":2},{"id":"fb7cfa6572df9a94c6f2f4ba5b27a6c695017e9f","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 14:31:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab5155b504819ff4590e345253e6002353ed600b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-06 14:39:34.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, and it does look like the staged import is a merge of lots of individual repos (I didn\u0027t check whether they\u0027re all represented but presumably Sunbeam project members will provide some cross-checking there).\n\nObviously hold approval on this until James weighs in as Sunbeam PTL.","accounts_in_message":[],"_revision_number":2},{"id":"b8b18d53382efbc8dc521fe3bba8091602ac89f2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-06 14:42:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4848c40c6708099df4058ec01a45c9a864cfa9d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-06 14:45:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c91d9ea1d56e42a6b88d02cfdb28b4f1079af624","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-06 14:49:13.000000000","message":"Patch Set 3: Code-Review+2\n\nGood catch!","accounts_in_message":[],"_revision_number":3},{"id":"ef2ed3126c62158430ea8bccfa2577c7004c0026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:58:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8fbc7d864d64bdb98b0cf3da010cc07\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bbe9695ea1df4a4489c7b4e8edfb65ff : SUCCESS in 5m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5ea23e0065b24639811059864660710a : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4953c306e345866ee9be2d54938f145463e2269c","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2023-11-07 05:59:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2209e958e41eed0c3aa9e9c0f1388ba1cf97adee","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-11-07 06:14:45.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d95fb783174ecfaa83a720cfc5c958a2c4d1156","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-10 16:16:15.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1417d4471511fa8f88f457505f1158391b88990d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:16:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ca22ada549462dd17affe1e9550d57db5253a2ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:23:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21845397c2b44bc790cb52361b83915e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f99e544dee242999dbcb00364fb2b2e : SUCCESS in 6m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b4964911664e43a8b49494a058692edb : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"3dacea588eb379c666a3ed38680a19744acba180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:23:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"324e47ce23b89cf7016db214d44fce4d1f848902","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:38:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e746b0203ee942e2b23d1366276b7fd4\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c411836cff244a68bb057f9e061d2eb0 : SUCCESS in 4m 42s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6062feb26c0f47b493ef7460ae1348a0 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4b11c4b99d73d40434f0b700deb96a6cda2bd4f6","revisions":{"469556727920ff027af7e58e234b221ee057825a":{"kind":"REWORK","_number":1,"created":"2023-11-06 11:29:46.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/67/900167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/900167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/900167/1"}}},"commit":{"parents":[{"commit":"87e56ff5fab49a5a4b78f0967142242933b6063b","subject":"Set proper LP project for ansible-collections-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87e56ff5fab49a5a4b78f0967142242933b6063b"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2023-11-06 11:11:21.000000000","tz":0},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2023-11-06 11:27:44.000000000","tz":0},"subject":"Create mono repo for sunbeam charms","message":"Create mono repo for sunbeam charms\n\nCurrently each sunbeam charm is in its own repo. This approach\nis going to be replaced by having a single monolithic repo that\nwill contain all the sunbeam charms.\n\nRequired-By: Ie72e31f8375371b87c97d9264aa4b92344efa0d9\nChange-Id: I79e5efca77a49b1cd2b0706d37a39cbbab4a19dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/469556727920ff027af7e58e234b221ee057825a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/469556727920ff027af7e58e234b221ee057825a"}]},"branch":"refs/heads/master"},"45e1f17ad6df400712146ef3086c6077769866da":{"kind":"REWORK","_number":2,"created":"2023-11-06 14:04:26.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/67/900167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/900167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/900167/2"}}},"commit":{"parents":[{"commit":"87e56ff5fab49a5a4b78f0967142242933b6063b","subject":"Set proper LP project for ansible-collections-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87e56ff5fab49a5a4b78f0967142242933b6063b"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2023-11-06 11:11:21.000000000","tz":0},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2023-11-06 14:04:12.000000000","tz":0},"subject":"Create mono repo for sunbeam charms","message":"Create mono repo for sunbeam charms\n\nCurrently each sunbeam charm is in its own repo. This approach\nis going to be replaced by having a single monolithic repo that\nwill contain all the sunbeam charms.\n\nRequired-By: Ie72e31f8375371b87c97d9264aa4b92344efa0d9\nChange-Id: I79e5efca77a49b1cd2b0706d37a39cbbab4a19dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45e1f17ad6df400712146ef3086c6077769866da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45e1f17ad6df400712146ef3086c6077769866da"}]},"branch":"refs/heads/master"},"4b11c4b99d73d40434f0b700deb96a6cda2bd4f6":{"kind":"REWORK","_number":3,"created":"2023-11-06 14:45:37.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/67/900167/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/900167/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/900167/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/900167/3"}}},"commit":{"parents":[{"commit":"87e56ff5fab49a5a4b78f0967142242933b6063b","subject":"Set proper LP project for ansible-collections-openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87e56ff5fab49a5a4b78f0967142242933b6063b"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2023-11-06 11:11:21.000000000","tz":0},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2023-11-06 14:45:26.000000000","tz":0},"subject":"Create mono repo for sunbeam charms","message":"Create mono repo for sunbeam charms\n\nCurrently each sunbeam charm is in its own repo. This approach\nis going to be replaced by having a single monolithic repo that\nwill contain all the sunbeam charms.\n\nRequired-By: Ie72e31f8375371b87c97d9264aa4b92344efa0d9\nChange-Id: I79e5efca77a49b1cd2b0706d37a39cbbab4a19dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b11c4b99d73d40434f0b700deb96a6cda2bd4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b11c4b99d73d40434f0b700deb96a6cda2bd4f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
