)]}'
{"id":"openstack%2Fproject-config~901916","triplet_id":"openstack%2Fproject-config~master~I61fd9b16f56d09bbc62cb1d602cfd927cd002a43","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-28 13:49:16.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-11-28 13:37:07.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I61fd9b16f56d09bbc62cb1d602cfd927cd002a43","subject":"Move os-(apply|collect|refresh)-config to heat","status":"MERGED","created":"2023-11-27 12:28:42.000000000","updated":"2023-11-28 13:59:39.000000000","submitted":"2023-11-28 13:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901916","meta_rev_id":"9535f206a32c07007b6813e0682d1919057735f3","_number":901916,"virtual_id_number":901916,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 13:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-28 13:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-28 13:37:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 12:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 12:39:26.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-27 16:19:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"bb5568e41baf22540ad5c1a5df8723f1fdaf4b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 12:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f835a5c48227b7f08da9c7e2dafcf0c509cd04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:36:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec090b469d8f4afdbfa524bf72331f1d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/70c7168263184ca3a89e3158daabf74f : SUCCESS in 7m 16s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/06ab05cf2b584df38f5d65cd26f12195 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e96a59deab7bdcf110226c33ce623ac6e0bebcc6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 12:37:05.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"59eb9cb12fa058bca9a0a548698a159af7800804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 12:39:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f3f6b8e6ad69260f7b177b86ade66b2744da5370","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-27 12:39:26.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            approval:\n              - Workflow: 1\n            require:\n              approval:\n                - Verified: [-1, -2]\n                  username: zuul\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"9d20fc92d9bde1b3cc2a65f50db0360297619b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:50:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d185d581ae514241ad73bd08e43385b5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bdf4014718bc4f8f96a072592b394cfb : SUCCESS in 5m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/728bbe5e101245c1b0c5ba6ba876a0e2 : SUCCESS in 3m 35s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/367090430e6a4ab4bee6399bc781ed62 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2777252604d6db940a0f053e5298cfca22c5bb01","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-27 16:19:44.000000000","message":"Patch Set 2: Code-Review-1\n\nIt looks like gerrit/acls/openstack/os-apply-config.config should also be cleaned up.","accounts_in_message":[],"_revision_number":2},{"id":"79e789caad171edb7637a3dfb9c2b80c8fbf6495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 01:06:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a9ee3cb6a0221da3277295adb23d1519dbdb1128","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-28 01:07:03.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            approval:\n              - Workflow: 1\n            require:\n              approval:\n                - Verified: [-1, -2]\n                  username: zuul\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":3},{"id":"c1fec849c21e906768bf30e5c5daed21b45f3700","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 01:07:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f028c8a96031f6622159b7bfa926f9f1641a8815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 01:22:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78704a884d7640b481334340359584d3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6e675da638134c5b9e3b5a05eb65d1a7 : SUCCESS in 6m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9ac6541ece564cb7974181b668c2ea1c : SUCCESS in 3m 38s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/d2f0fa290315421b935d82aff68d6308 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9d619dcc3855f46bf2a2cb15b7f7954173bd90b8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-28 13:37:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThe reply weirdness is probably because I\u0027m using the GerTTY client, and it uses an older API call to add patchset level comments.","accounts_in_message":[],"_revision_number":3},{"id":"7417b801db189cce4036697dea493b9aaae73811","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:37:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bad8d68709a869b644256080fb2c0e8fa9b9834b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:49:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0a8771ab124a50ad34ab7e85b4b0d3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/112c4bdc313f4e1aa44521aec43ab3e2 : SUCCESS in 6m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d3b2e804ebce47aca0f60adf9b133393 : SUCCESS in 3m 43s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/5fab3de1d5e14587bf6d882f4e0e17ce : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"2b94cb4500a63e19ac1a88c12f73d552a26d29a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:49:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9535f206a32c07007b6813e0682d1919057735f3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:59:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e03b4ffd3b80465dacdc51c60af0e5f0\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0976cac1960945599d63f783ecc72104 : SUCCESS in 4m 08s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/404b9c7251794eefb8ae267719883e4d : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"85294041b121e803fa3ebac696541ccb13eb0a0c","revisions":{"9ed153a580c727a722cb7faee1a41374f7defd2c":{"kind":"REWORK","_number":1,"created":"2023-11-27 12:28:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/901916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/901916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/901916/1"}}},"commit":{"parents":[{"commit":"7fa97b142a105b185202363ebe9ef860cbc7f9a0","subject":"Add new Gerrit submodule java-prettify to Zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fa97b142a105b185202363ebe9ef860cbc7f9a0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 12:27:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 12:28:39.000000000","tz":540},"subject":"Redirect irc notifications from os-(apply|collect|refresh)-config","message":"Redirect irc notifications from os-(apply|collect|refresh)-config\n\nThese projects have been transferred from tripleo to heat[1].\n\n[1] 55873c8fcb825611d17b60e61f1c66d2981e39b3\n\nChange-Id: I61fd9b16f56d09bbc62cb1d602cfd927cd002a43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ed153a580c727a722cb7faee1a41374f7defd2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ed153a580c727a722cb7faee1a41374f7defd2c"}]},"branch":"refs/heads/master"},"4070ed52908363a47790cfa03c967306dd360d47":{"kind":"REWORK","_number":2,"created":"2023-11-27 12:39:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/901916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/901916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/901916/2"}}},"commit":{"parents":[{"commit":"7fa97b142a105b185202363ebe9ef860cbc7f9a0","subject":"Add new Gerrit submodule java-prettify to Zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fa97b142a105b185202363ebe9ef860cbc7f9a0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 12:27:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 12:38:30.000000000","tz":540},"subject":"Move os-(apply|collect|refresh)-config to heat","message":"Move os-(apply|collect|refresh)-config to heat\n\nThese projects have been transferred from tripleo to heat[1].\n\nAlso replace the os-apply-config group because the group contains only\nTripleO cores.\n\n[1] 55873c8fcb825611d17b60e61f1c66d2981e39b3\n\nChange-Id: I61fd9b16f56d09bbc62cb1d602cfd927cd002a43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4070ed52908363a47790cfa03c967306dd360d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4070ed52908363a47790cfa03c967306dd360d47"}]},"branch":"refs/heads/master"},"85294041b121e803fa3ebac696541ccb13eb0a0c":{"kind":"REWORK","_number":3,"created":"2023-11-28 01:06:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/901916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/901916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/901916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/901916/3"}}},"commit":{"parents":[{"commit":"7fa97b142a105b185202363ebe9ef860cbc7f9a0","subject":"Add new Gerrit submodule java-prettify to Zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fa97b142a105b185202363ebe9ef860cbc7f9a0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 12:27:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 01:06:14.000000000","tz":540},"subject":"Move os-(apply|collect|refresh)-config to heat","message":"Move os-(apply|collect|refresh)-config to heat\n\nThese projects have been transferred from tripleo to heat[1].\n\nAlso replace the os-apply-config group because the group contains only\nTripleO cores.\n\n[1] 55873c8fcb825611d17b60e61f1c66d2981e39b3\n\nChange-Id: I61fd9b16f56d09bbc62cb1d602cfd927cd002a43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/85294041b121e803fa3ebac696541ccb13eb0a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/85294041b121e803fa3ebac696541ccb13eb0a0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
