)]}'
{"id":"openstack%2Fproject-config~907060","triplet_id":"openstack%2Fproject-config~master~I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"35904":{"account":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"last_update":"2024-01-31 13:17:55.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2024-01-31 09:00:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-01-31 13:09:59.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7","subject":"Add new components to NebulOuS project: prediction-orchestrator, exn-middleware, overlay-network-agent","status":"MERGED","created":"2024-01-29 13:48:51.000000000","updated":"2024-01-31 13:30:16.000000000","submitted":"2024-01-31 13:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907060","meta_rev_id":"063560c9c1a7fedca2861d4313f16738fd8b9543","_number":907060,"virtual_id_number":907060,"owner":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-31 13:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2024-01-31 09:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-31 13:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-31 13:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-29 13:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-29 13:55:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2024-01-30 13:47:14.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-01-31 08:37:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"a897e361b980eeeab7d986f5bb6145808856e900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-01-29 13:48:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c537f30ccead52fc4868287f962d1365633a0860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 13:55:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/762c68a5ae2640b39d11b3dbe5c43556\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/238922ff39074a62b3f00a478678b0f3 : SUCCESS in 5m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dd9699de40e34ab388400409ee7399e9 : FAILURE in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"26430caa47192bc664bb569fd6e3dc53a74a9ec3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-29 13:55:54.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nSee comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"9a78b31eb8502d851a648eeba086efd21131a49d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-01-29 17:13:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a2e1df7b535bf643de84e2b078ad02b20db406b5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-29 17:18:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI\u0027ve noted a new (minor) concern inline now.","accounts_in_message":[],"_revision_number":2},{"id":"55932cb586222f5835d958031c16b71def61d300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 17:30:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/081defcd88a848b59b8f8390bb2debaf\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/467d8ccd1e184cd685b06c2a09f687bb : SUCCESS in 7m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c6cd96b3a65a4b089a9c3d7293eb028d : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a090b7ea8f681cff89c4a233e0aa311116ee40a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-01-30 09:54:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"da897594945f384d0a26807240757cd35e132166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 10:02:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/873ad7e03b104803b773947755077cfd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3aca32f55a4b44a68a6c60fd2a066c04 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cc573d5055064f0fa29789f1cd9cc7f7 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"aa2e6d10f12803e58ca6401cde405373e46a4152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-01-30 10:12:45.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"aff99578e79ce1a1d1c0f8e5888e783be8f87177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 10:21:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbb2273035204ff7abe5bef286735310\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/24130972057d49f3a803caea133e963d : SUCCESS in 7m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2b94a36c73cd44aeb36cdab3c237d38b : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":4},{"id":"677909b0148a5dec5ee8bbc2ab4fed91273cad93","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-30 13:47:14.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM but let\u0027s make sure Radosław is aware before we approve this.","accounts_in_message":[],"_revision_number":4},{"id":"744be7248e3203e805e3b0018ddd95ce67c8ec55","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-01-31 08:37:59.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5de9c60147d53d04bf91d5dc3b7a97009fbe5bb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-01-31 08:50:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2d0f5f9a75d16479627c3b2bf271602fbb976a95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-01-31 09:00:28.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"55f5d80d01a9969bce06440cafa408f093d5fce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 09:02:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f290117618c34959a7085e39d8b6dfee\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/22dcc947d5024e5fb257a74c6cf05dcc : SUCCESS in 6m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aac7fb1ac6e544989747c07796dd6e1d : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"9c44482e8676d81925375edb252de9a8acc47c0b","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-01-31 09:06:56.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"65ce9dc5fde174f00595e02ba79538b2026cdee3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-31 13:09:59.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"936c452ee69b4c566e156847e095b0da737493c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:10:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5b22d756ddd0a6e999f09d1e47e0c27dd257726b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:17:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a874b187c4df4462967778fb26cb6427\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/43182156fe8748618d4995d0526d67e0 : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/252f9c306b944fea9b0d39f4200ea79f : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":5},{"id":"1f3db95957066821e4a0b31cd8e17ee93dd5c395","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:17:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"063560c9c1a7fedca2861d4313f16738fd8b9543","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:30:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1463874844174c95a61b7bc8c17b18b0\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/633c624a360c48c28041df15d08ab573 : SUCCESS in 3m 59s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/a76efbfeef1b4b6198117a39691973f3 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c7e553c395fc5ebd2263fac023df06705f369fb","revisions":{"3d7e834e88cf6fd4975b54ae662cb110227df280":{"kind":"REWORK","_number":1,"created":"2024-01-29 13:48:51.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/60/907060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/907060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/907060/1"}}},"commit":{"parents":[{"commit":"f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1","subject":"Merge \"Update secret for Weblate REST API key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 13:43:59.000000000","tz":60},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 13:48:38.000000000","tz":60},"subject":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent","message":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent\n\nChange-Id: I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d7e834e88cf6fd4975b54ae662cb110227df280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d7e834e88cf6fd4975b54ae662cb110227df280"}]},"branch":"refs/heads/master"},"9cb76050370e886578c0ea4d09e9ad5876d92234":{"kind":"REWORK","_number":2,"created":"2024-01-29 17:13:52.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/60/907060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/907060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/907060/2"}}},"commit":{"parents":[{"commit":"f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1","subject":"Merge \"Update secret for Weblate REST API key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 13:43:59.000000000","tz":60},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 17:13:42.000000000","tz":60},"subject":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent","message":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent\n\nChange-Id: I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cb76050370e886578c0ea4d09e9ad5876d92234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cb76050370e886578c0ea4d09e9ad5876d92234"}]},"branch":"refs/heads/master"},"22897d2fbe6bd21a055bab8c4269731ee41aa0bf":{"kind":"REWORK","_number":3,"created":"2024-01-30 09:54:14.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/60/907060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/907060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/907060/3"}}},"commit":{"parents":[{"commit":"f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1","subject":"Merge \"Update secret for Weblate REST API key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 13:43:59.000000000","tz":60},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-30 09:54:05.000000000","tz":60},"subject":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent","message":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent\n\nChange-Id: I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22897d2fbe6bd21a055bab8c4269731ee41aa0bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22897d2fbe6bd21a055bab8c4269731ee41aa0bf"}]},"branch":"refs/heads/master"},"7371bd434dc4a88c29c5d9c31632c1f6c9e8422e":{"kind":"NO_CHANGE","_number":4,"created":"2024-01-30 10:12:45.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/60/907060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/907060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/907060/4"}}},"commit":{"parents":[{"commit":"f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1","subject":"Merge \"Update secret for Weblate REST API key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 13:43:59.000000000","tz":60},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-30 10:12:33.000000000","tz":60},"subject":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent","message":"Add new components to NebulOuS project: prediction-orchiestrator, exn-middleware, overlay-network-agent\n\nChange-Id: I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7371bd434dc4a88c29c5d9c31632c1f6c9e8422e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7371bd434dc4a88c29c5d9c31632c1f6c9e8422e"}]},"branch":"refs/heads/master"},"6c7e553c395fc5ebd2263fac023df06705f369fb":{"kind":"REWORK","_number":5,"created":"2024-01-31 08:50:36.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/60/907060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/907060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/907060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/907060/5"}}},"commit":{"parents":[{"commit":"f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1","subject":"Merge \"Update secret for Weblate REST API key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6d95b6f27d2bc5eb8a33c2dd6f23849898983f1"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-29 13:43:59.000000000","tz":60},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-01-31 08:50:17.000000000","tz":60},"subject":"Add new components to NebulOuS project: prediction-orchestrator, exn-middleware, overlay-network-agent","message":"Add new components to NebulOuS project: prediction-orchestrator, exn-middleware, overlay-network-agent\n\nChange-Id: I27ed3c7427b2f4bbfbe0d23ee9a5fca8b606efc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c7e553c395fc5ebd2263fac023df06705f369fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6c7e553c395fc5ebd2263fac023df06705f369fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
