)]}'
{"id":"openstack%2Fproject-config~907150","triplet_id":"openstack%2Fproject-config~master~I7f4104a342d170669e7717b945f09131811dbebc","project":"openstack/project-config","branch":"master","topic":"fix-centos-stream-wheel-mirror","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-01-30 05:05:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7f4104a342d170669e7717b945f09131811dbebc","subject":"Fix wheel_volume values for centos stream wheel mirrors","status":"MERGED","created":"2024-01-29 22:10:12.000000000","updated":"2024-01-30 05:05:15.000000000","submitted":"2024-01-30 05:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907150-fix-centos-stream-wheel-mirror","meta_rev_id":"ffe7532b28e8db0bec28549c20a32f9475faac8c","_number":907150,"virtual_id_number":907150,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-30 05:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-30 04:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2024-01-29 22:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-30 04:59:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-29 22:10:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-29 22:18:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-29 22:21:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-01-30 04:59:00.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"e8c76925f894a2d419987bfe83208360664df7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-01-29 22:10:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd6f18b2c3d53252b09e5c526af98369830869bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-29 22:10:36.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            approval:\n              - Workflow: 1\n            require:\n              approval:\n                - Verified: [-1, -2]\n                  username: zuul\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ef6e42f91be35f8476f51000ff922ae2bae483ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 22:18:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb54748afa2494693804b442ff267a3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a7337e911047462b8d37c30ba4fcf333 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f82f2703ea2ed68c5bebb436b845370cb54d6e91","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-29 22:21:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ab627cb62bb2d27aa09bd6eaabb204747db9f96","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2024-01-30 04:59:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a40fe00765772aac721fcee14ed47501ae79958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 04:59:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf1f8cce666c7850a3477cd4dadafd63fff0d120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 05:05:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7844c4df2341909116a70b57400d85\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ed399f09b710477c9c4c8b361f6d235c : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ffe7532b28e8db0bec28549c20a32f9475faac8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 05:05:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21a2d0dcb1db3ed439559ffe451e011b93914ea7","revisions":{"21a2d0dcb1db3ed439559ffe451e011b93914ea7":{"kind":"REWORK","_number":1,"created":"2024-01-29 22:10:12.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/50/907150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/907150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/907150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/907150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/907150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/907150/1"}}},"commit":{"parents":[{"commit":"4b5f31b9bb5d4895e3a0ec463ad9f49c12bd6794","subject":"Merge \"Implement ironic-unmaintained-core group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b5f31b9bb5d4895e3a0ec463ad9f49c12bd6794"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-01-29 22:08:54.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-01-29 22:08:54.000000000","tz":-480},"subject":"Fix wheel_volume values for centos stream wheel mirrors","message":"Fix wheel_volume values for centos stream wheel mirrors\n\nWe were trying to vos release the non stream centos 8 volumes which\ndon\u0027t exist anymore. Flip this to the stream values. Additionally,\ncentos 9 stream arm64 was releasing x86_64 wheels not arm64 wheels. Fix\nthis too.\n\nThis probably is an indication that these mirrors were never really used\nmuch in the first place and we should consider cleanup?\n\nChange-Id: I7f4104a342d170669e7717b945f09131811dbebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21a2d0dcb1db3ed439559ffe451e011b93914ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21a2d0dcb1db3ed439559ffe451e011b93914ea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
