)]}'
{"id":"openstack%2Fproject-config~907362","triplet_id":"openstack%2Fproject-config~master~I1262b0c9ef203d4408d74776b2d8b3ffaa0c8761","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I1262b0c9ef203d4408d74776b2d8b3ffaa0c8761","subject":"DNM: test syntax error","status":"ABANDONED","created":"2024-01-31 21:44:44.000000000","updated":"2024-01-31 23:21:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ae0580116ce312263eed5ebdc73314e87c5ece4","_number":907362,"virtual_id_number":907362,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-31 21:50:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 21:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f912c413e3273b47d575f88a769c76860f3a674b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-01-31 21:44:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1c3766e1e17f945edf744c4c74e331db9066a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 21:50:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95cfe316236941ca95adc6324b187d5b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f78fae2cc7854d5ba158e8a28948d410 : SUCCESS in 5m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cff827774e844766ba74782e5f878455 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9ae0580116ce312263eed5ebdc73314e87c5ece4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-01-31 23:21:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"654f689af6ecaf5f0eb1463137bdd0fb579beebe","revisions":{"654f689af6ecaf5f0eb1463137bdd0fb579beebe":{"kind":"REWORK","_number":1,"created":"2024-01-31 21:44:44.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/62/907362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/907362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/907362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/907362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/907362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/907362/1"}}},"commit":{"parents":[{"commit":"6bbe1ab0e5223f82c847306c354992e1c434bb72","subject":"Merge \"Retire oswin-tempest-plugin: remove project from infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6bbe1ab0e5223f82c847306c354992e1c434bb72"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-01-31 21:44:36.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-01-31 21:44:36.000000000","tz":-480},"subject":"DNM: test syntax error","message":"DNM: test syntax error\n\nChange-Id: I1262b0c9ef203d4408d74776b2d8b3ffaa0c8761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/654f689af6ecaf5f0eb1463137bdd0fb579beebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/654f689af6ecaf5f0eb1463137bdd0fb579beebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
