)]}'
{"id":"openstack%2Fproject-config~909445","triplet_id":"openstack%2Fproject-config~master~Ic89b9b1f51934f6a05d11616b7adba86684edcf2","project":"openstack/project-config","branch":"master","topic":"charm-stable-maint-ovn","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2024-04-10 15:29:10.000000000","reason":"Change was submitted"},"20870":{"account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"last_update":"2024-04-10 15:29:10.000000000","reason":"Change was submitted"},"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2024-04-10 08:46:40.000000000","reason":"\u003cGERRIT_ACCOUNT_13686\u003e replied on the change","reason_account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"}},"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2024-02-27 21:04:57.000000000","reason":"\u003cGERRIT_ACCOUNT_2424\u003e replied on the change","reason_account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}}},"hashtags":[],"change_id":"Ic89b9b1f51934f6a05d11616b7adba86684edcf2","subject":"Add ovn charms to charms-stable-maint purview","status":"MERGED","created":"2024-02-19 12:30:13.000000000","updated":"2024-04-10 15:33:14.000000000","submitted":"2024-04-10 15:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909445-charm-stable-maint-ovn","meta_rev_id":"b90d4658c69f36477878c587d3a721e92a1e4697","_number":909445,"virtual_id_number":909445,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-10 15:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"all":[{"value":1,"date":"2024-04-10 08:47:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2024-02-27 21:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-27 21:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-10 15:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}],"reviewers":{"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-19 12:31:15.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"CC"},{"updated":"2024-02-19 12:31:15.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-02-19 12:31:15.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"REVIEWER"},{"updated":"2024-02-19 12:31:15.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2024-02-19 12:47:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-19 13:04:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2024-02-26 16:49:07.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"},{"updated":"2024-02-27 21:09:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e6406ae05f1e5779f69045bc1dde2b406e85df11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2024-02-19 12:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a631074d0ab83083305a59ef29f236276f5a72a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 12:47:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3576451f9ff54ab2a0359f69f1ee5da5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e765654edf664d358167a9e9bef49ee4 : SUCCESS in 6m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a14e5442c7be4e88badd7c3f2b9e0e17 : FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2ae1b38ec9b59044e01c0887e2ecbd979b489f95","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-19 13:04:38.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nIt\u0027s unusual for an official OpenStack team\u0027s groups to appear in ACLs for repositories outside the \"openstack/\" Git namespace. Would it make sense to add a separate ovn-charms-stable-maint group and then that group can have the charms-stable-maint group added as an included group for convenience?","accounts_in_message":[],"_revision_number":1},{"id":"8d3105094fa1a1dc1b29f4e7908236b19db10e26","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2024-02-26 16:48:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d48c1976d3ce472ebcc80ac5dd20609ddcd1244e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2024-02-26 16:48:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"69345ce12fc8fe9686c952bdd46d0d3feeffe5d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 17:02:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f62c29ef4ea44feda8f9c227a835f29f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/df1bd070381240258d3dd51523af12bd : SUCCESS in 7m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d5837e418d6a463f9719051597eb28ea : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"22be516c57d83a4339eb04299d1bdc828e63a746","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2024-02-27 21:04:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f9a09a7289910b2e8eee12c0ee5922521006678","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-27 21:08:16.000000000","message":"Patch Set 2:\n\nIf \"the ovn charms are actually part of the same charm set as the openstack charms\" then why not officially adopt these repositories as deliverables of the OpenStack Charms team? I guess it\u0027s not clear at all why these are being kept out of OpenStack while still maintained by an official OpenStack project team.","accounts_in_message":[],"_revision_number":2},{"id":"9f9c68559de173550b3779d7e7a8f4ec21534a32","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-27 21:09:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"82939b8bc9b658608aaf42eff0ce54b676427315","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2024-03-21 09:35:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3f483f4f8fc6d9791be8a1756b21dc7b316084f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-21 12:20:17.000000000","message":"Patch Set 2:\n\nI hadn\u0027t seen an answer to my earlier question. I just want to make sure we\u0027re not maintaining shadow components of OpenStack Charms outside the openstack/ Git namespace.","accounts_in_message":[],"_revision_number":2},{"id":"8ffea6b204767ac48a78809fa1c832446a7007e3","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2024-04-10 08:46:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f578e7ac9dcbb5251dee68c4e42ba9e61e8ecffc","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2024-04-10 08:47:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4fb93b5f53b39a90313744ddb6b44891cf35422f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-10 15:18:31.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5fb92521f560413e8328a20eb7da690f41a49e7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 15:18:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"476859cc2795687902b97826207212bffa620ace","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 15:29:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8693f8648524ccaa3c375e1f8bca847\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8a29fdf9cba42909818d48efe057d02 : SUCCESS in 5m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cad669655a334ccfaba44507a984934a : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4a081e113e3c9185ded6c383f445de1ba37786d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 15:29:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b90d4658c69f36477878c587d3a721e92a1e4697","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 15:33:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4a3891b3494129b817ae6f2a4f76e8\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c530dc9974ec4d93b575a57e0ec8d364 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab04717eadf2984c3739162961c78a96123090ae","revisions":{"ff898bb87880d8e8bb925cb7aeff6b136858a63a":{"kind":"REWORK","_number":1,"created":"2024-02-19 12:30:13.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/45/909445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/909445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/909445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/909445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/909445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/909445/1"}}},"commit":{"parents":[{"commit":"6d97f3331f7197104e10dfc192f19114361b7878","subject":"Merge \"End gate and update acl for retiring openstack-chef projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d97f3331f7197104e10dfc192f19114361b7878"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2024-02-19 12:27:30.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2024-02-19 12:27:30.000000000","tz":0},"subject":"Add ovn charms to charms-stable-maint purview","message":"Add ovn charms to charms-stable-maint purview\n\nThe charms-stable-maint group has maintenance rights\nof the Openstack charms stable branches. This patch\nwidens this set of charms to include the OVN charms\nstable branches.\n\nChange-Id: Ic89b9b1f51934f6a05d11616b7adba86684edcf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff898bb87880d8e8bb925cb7aeff6b136858a63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff898bb87880d8e8bb925cb7aeff6b136858a63a"}]},"branch":"refs/heads/master"},"ab04717eadf2984c3739162961c78a96123090ae":{"kind":"REWORK","_number":2,"created":"2024-02-26 16:48:37.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/45/909445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/909445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/909445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/909445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/909445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/909445/2"}}},"commit":{"parents":[{"commit":"6d97f3331f7197104e10dfc192f19114361b7878","subject":"Merge \"End gate and update acl for retiring openstack-chef projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d97f3331f7197104e10dfc192f19114361b7878"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2024-02-19 12:27:30.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2024-02-26 16:47:15.000000000","tz":0},"subject":"Add ovn charms to charms-stable-maint purview","message":"Add ovn charms to charms-stable-maint purview\n\nThe charms-stable-maint group has maintenance rights\nof the Openstack charms stable branches. This patch\nwidens this set of charms to include the OVN charms\nstable branches.\n\nChange-Id: Ic89b9b1f51934f6a05d11616b7adba86684edcf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab04717eadf2984c3739162961c78a96123090ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab04717eadf2984c3739162961c78a96123090ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
