)]}'
{"id":"openstack%2Fproject-config~910011","triplet_id":"openstack%2Fproject-config~master~Iaab540fdf67f357080bed7aff62f7136596fa8fd","project":"openstack/project-config","branch":"master","topic":"drop-buster","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-02-24 16:06:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaab540fdf67f357080bed7aff62f7136596fa8fd","subject":"Drop Debian Buster python wheel mirror publication jobs","status":"MERGED","created":"2024-02-23 16:42:29.000000000","updated":"2024-02-24 16:06:48.000000000","submitted":"2024-02-24 16:06:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910011-drop-buster","meta_rev_id":"f8ad136edf99c74703c655fe8f51d014e7a22bc5","_number":910011,"virtual_id_number":910011,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-24 16:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-24 15:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-24 16:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 16:46:21.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-23 16:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-24 15:57:04.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"3d4f848c31d3add191c43fea0296f68f93c71ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-02-23 16:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2476760b0f34577de8666c0792545d015c330ed6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-23 16:46:21.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            approval:\n              - Workflow: 1\n            require:\n              approval:\n                - Verified: [-1, -2]\n                  username: zuul\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"7f5a777a57263e290373ad521c35fcf24a321325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 16:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e707480f3b06427ab174f2a2fe141ec5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2fc9997f1f2f49f08ed2d05ee981752e : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d946f4d385651e0c71d4b7ab3e4fe078a3af6f34","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-24 15:57:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a005f57fd2b2684546e58fa1f355b767c97dfc76","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-24 16:01:03.000000000","message":"Patch Set 1: Workflow+1\n\nApproving to un-block the related changes complaining about invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"f3a802eaabdcc43ad0f47c3b0a7e6d67b6613f1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:01:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"128aed5dba8da9d509d6e3c212e33067d06b7785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:06:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c14b24f6b374ca6afbe9d9ea17663f6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8b53568dade14dd1ae4015157bb28910 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f8ad136edf99c74703c655fe8f51d014e7a22bc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 16:06:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acf5eedb2ccdd9aa235628251d4c69f4d09b1e8e","revisions":{"acf5eedb2ccdd9aa235628251d4c69f4d09b1e8e":{"kind":"REWORK","_number":1,"created":"2024-02-23 16:42:29.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/11/910011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/11/910011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/910011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/910011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/11/910011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/11/910011/1"}}},"commit":{"parents":[{"commit":"211fe14946a97536d4e60a8e2c5c8b22ab8aada6","subject":"Remove opensuse-15 image builds from nodepool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/211fe14946a97536d4e60a8e2c5c8b22ab8aada6"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-02-23 16:39:46.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-02-23 16:39:46.000000000","tz":-480},"subject":"Drop Debian Buster python wheel mirror publication jobs","message":"Drop Debian Buster python wheel mirror publication jobs\n\nWe are in the process of dropping this node type from nodepool and our\nmirrors entirely. Stop running jobs to build the python wheel\nmirror/cache for this release.\n\nChange-Id: Iaab540fdf67f357080bed7aff62f7136596fa8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/acf5eedb2ccdd9aa235628251d4c69f4d09b1e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/acf5eedb2ccdd9aa235628251d4c69f4d09b1e8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
