)]}'
{"id":"openstack%2Fproject-config~910270","triplet_id":"openstack%2Fproject-config~master~Ib2c64b65a7655b92c2b3afeff6db292eaf2212d4","project":"openstack/project-config","branch":"master","topic":"retire-tripleo","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-28 08:31:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2c64b65a7655b92c2b3afeff6db292eaf2212d4","subject":"Retire TripleO: remove project from infra","status":"MERGED","created":"2024-02-26 20:28:42.000000000","updated":"2024-02-28 08:44:56.000000000","submitted":"2024-02-28 08:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910270-retire-tripleo","meta_rev_id":"840782db80216f1fae9fdcd254238888f42d5016","_number":910270,"virtual_id_number":910270,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 08:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-28 08:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-02-28 07:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-28 08:20:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 20:29:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 20:36:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-28 07:42:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-28 08:20:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"86173a57758a16e72f9bc168e7bc40fc311a95f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-26 20:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41e0a79e25bee4f0f67e68ab14558f99177ff0bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-26 20:29:01.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            approval:\n              - Workflow: 1\n            require:\n              approval:\n                - Verified: [-1, -2]\n                  username: zuul\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"70d822b8e229ba9c402dfa37c912d2231f7aa565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 20:36:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2cde1ee48746e5834d471e92f68805\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0d036ae587fa442f86fdfa1a382e95a9 : SUCCESS in 4m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1cd0291d241e4121ae49861abc73be42 : SUCCESS in 4m 31s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/fd402ec404cb4580aa910636059a6d25 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6eabcf07c7734af96834465f52d3132f142d17d7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-28 07:42:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5924a3a3271ea53ac46d0ce98c24324725c1d2c7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-28 08:20:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fd55632ec12bd748aaac8d7c27f9e39303a98f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"585f2e4a57c281ab13ac11d6f0cd052c91bbdc27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2fff3915b74acdb70afbd00745555f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b9fa3fdc528d40d4a20ea283561800f6 : SUCCESS in 6m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0cb1395366f945558bfd23caf6bdcdd0 : SUCCESS in 4m 17s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e244ab8515964ac5ad13f1dae885ace6 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a00917d1f3d6a03cd02e6a95a5f85652088a091b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"840782db80216f1fae9fdcd254238888f42d5016","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:44:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d009a271f15456ca8ca26293a44e191\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/78f68b0f67a24f6da29983255c820588 : SUCCESS in 5m 37s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/d77f19f08c3f437d8ec619065b2da622 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"884f77006171216abdb7331e66d033bcfe026dc9","revisions":{"884f77006171216abdb7331e66d033bcfe026dc9":{"kind":"REWORK","_number":1,"created":"2024-02-26 20:28:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/910270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/70/910270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/910270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/910270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/70/910270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/70/910270/1"}}},"commit":{"parents":[{"commit":"3adfb1106b49f33c91ffd240fcc1c71e5da3877d","subject":"Merge \"Set debian-buster min servers to 0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3adfb1106b49f33c91ffd240fcc1c71e5da3877d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-26 20:25:02.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-26 20:25:02.000000000","tz":-480},"subject":"Retire TripleO: remove project from infra","message":"Retire TripleO: remove project from infra\n\nTripleO project is retiring\n- https://review.opendev.org/c/openstack/governance/+/905145\n\nthis commit removes it from infra except trieplo-ci repo\nwhich needs to be removed from infra at the end. We need to\nremove other repo (user of jobs defined in tripleo-ci repo)\nfirst so tripleo-ci repo content can be cleanup.\n\nChange-Id: Ib2c64b65a7655b92c2b3afeff6db292eaf2212d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/884f77006171216abdb7331e66d033bcfe026dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/884f77006171216abdb7331e66d033bcfe026dc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
