)]}'
{"id":"openstack%2Fproject-config~910580","triplet_id":"openstack%2Fproject-config~master~I7e6e0899aa783e8abf5531f2ff67e81be08991cd","project":"openstack/project-config","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-03-01 14:37:08.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I7e6e0899aa783e8abf5531f2ff67e81be08991cd","subject":"Add OpenAPI related repos to OpenStackSDK project","status":"MERGED","created":"2024-02-29 08:32:45.000000000","updated":"2024-03-28 10:14:34.000000000","submitted":"2024-03-01 14:39:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910580-openapi","meta_rev_id":"b6c6fcca19c68da29756bbcb6e82e43ed00fa1f3","_number":910580,"virtual_id_number":910580,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 14:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-01 14:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-01 14:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 08:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-01 14:32:41.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-03-28 10:14:34.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"d767ca73c2150c18f4473ddb1f6f194f9d5fc0e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-02-29 08:32:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1f9fd9570984e71289ad6158168ce18a45c157e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-02-29 08:33:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"30dc7431c91d41c591fd544f82cedd5216e579c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 08:43:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6183fe48b7c644d9bfbbda117e94691d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/309c4a163f3c47c5ab7a9db29b4bd69b : SUCCESS in 6m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/230c0ebab73645c89ba3a5e8d16e328e : SUCCESS in 3m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/16f3f1450a4746c3b005f21ac9b26076 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7220b8e5f6b1c28359e1a05e36664bb6a03eaf86","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-01 14:32:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa9622c14ac2d45fd725e8f990985ec510c9bfab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 14:33:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da7ab666a5348fc1417a6278579b1632bbc146e8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-03-01 14:37:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9060bd47a5834dd76f7413c6eea9e3dec003337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 14:39:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01e28e73b91c40369117c0ad6c2a7b8f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/68aa8a7865ab40e5b98cf55a6ba35411 : SUCCESS in 5m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e9c82f8c2385423badf62a7cdbb214e8 : SUCCESS in 3m 42s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/292a1f5b51534e54b5bd5865b9bf66a2 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"87b9707c8ceee66eec8c112f8f28c4cdf10979d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 14:39:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5175272cc468dc76a62dee1ec55b00bd8218284a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 14:52:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69946a23a5cd43a18f9d2c1ea3d70c48\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f2a4b16eb56c415e97af83e5970a9705 : SUCCESS in 4m 03s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/790b360ee8ef4011bcfd23949d10286a : SUCCESS in 5m 13s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/acfd94b080f7464da8b92a7e54eef88f : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b9c29469477fc6a4d87384fb4f6b6ffc772aa3e","revisions":{"5f221de7457cce6636c4f97e859e6d7c181d5a9f":{"kind":"REWORK","_number":1,"created":"2024-02-29 08:32:45.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/80/910580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/910580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/910580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/910580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/910580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/910580/1"}}},"commit":{"parents":[{"commit":"224e109f11462145859f238976d136df72a5a365","subject":"Merge \"Remove debian-buster image builds from nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/224e109f11462145859f238976d136df72a5a365"}]}],"author":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-29 08:29:50.000000000","tz":60},"committer":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-29 08:29:50.000000000","tz":60},"subject":"Add OpenAPI related repos to OpenStackSDK project","message":"Add OpenAPI related repos to OpenStackSDK project\n\n- add `codegenerator` repository used to generate OpenAPI specs from\n  service sources and client API bindings from the OpenAPI\n- add `openapi` repository with documents describing OpenStack\n  flavouring of OpenAPI and Sphinx extension to render them similar to\n  os-api-ref\n\nChange-Id: I7e6e0899aa783e8abf5531f2ff67e81be08991cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f221de7457cce6636c4f97e859e6d7c181d5a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f221de7457cce6636c4f97e859e6d7c181d5a9f"}]},"branch":"refs/heads/master"},"8b9c29469477fc6a4d87384fb4f6b6ffc772aa3e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-29 08:33:57.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/80/910580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/910580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/910580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/910580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/910580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/910580/2"}}},"commit":{"parents":[{"commit":"224e109f11462145859f238976d136df72a5a365","subject":"Merge \"Remove debian-buster image builds from nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/224e109f11462145859f238976d136df72a5a365"}]}],"author":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-29 08:29:50.000000000","tz":60},"committer":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2024-02-29 08:33:40.000000000","tz":60},"subject":"Add OpenAPI related repos to OpenStackSDK project","message":"Add OpenAPI related repos to OpenStackSDK project\n\n- add `codegenerator` repository used to generate OpenAPI specs from\n  service sources and client API bindings from the OpenAPI\n- add `openapi` repository with documents describing OpenStack\n  flavouring of OpenAPI and Sphinx extension to render them similar to\n  os-api-ref\n\nNeeded-By: https://review.opendev.org/c/openstack/governance/+/910581\n\nChange-Id: I7e6e0899aa783e8abf5531f2ff67e81be08991cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b9c29469477fc6a4d87384fb4f6b6ffc772aa3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b9c29469477fc6a4d87384fb4f6b6ffc772aa3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
