)]}'
{"id":"openstack%2Fproject-config~911381","triplet_id":"openstack%2Fproject-config~master~I97887063c735c96d200ce2cbd8950bbec0ef7240","project":"openstack/project-config","branch":"master","topic":"rackspace-mfa","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-03-06 16:40:34.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-03-07 14:00:32.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I97887063c735c96d200ce2cbd8950bbec0ef7240","subject":"Try switching Rackspace DFW to an API key","status":"MERGED","created":"2024-03-05 22:19:05.000000000","updated":"2024-03-07 14:38:55.000000000","submitted":"2024-03-07 14:26:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911381-rackspace-mfa","meta_rev_id":"3074684e2b24947e3c14751c09b166e93f1cb8fe","_number":911381,"virtual_id_number":911381,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 14:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-03-06 16:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-07 14:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 22:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 23:21:37.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"819c6bcb302da85f559d4dc09a621430ca743839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-05 22:19:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06b932530f94a46376f7cce043f47b61ed216fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 22:24:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/477c64f4f7b941569563327ec0533335\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aa51bd3f0cba4715952e299002071278 : SUCCESS in 4m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d74783c16a6e47b886c0dbd51177ecd7 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"51f102e1ea8e511bafe41b23456ebd5b2302ccb2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-05 23:21:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e60be5ac15e6ff7fad3e8db3d66a4366f4b0ae57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-06 15:06:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0daef17fd50484b38d07b471dcfa3ab07457c836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:14:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9beaa369f23443e6a95ca6c1ac94d5b6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5fc9ad81867442c69d8dd99d39fc1750 : SUCCESS in 5m 14s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/cb40451a827645a2a2bba194793ff2b9 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d3c1f51bf29013357c396b2b64ecbdeeb1cd3e38","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-06 16:40:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"947be9bcbd8f377c98c25c41bf05af7dec354c48","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-07 14:00:32.000000000","message":"Patch Set 2: Workflow+1\n\nI\u0027ll test this on nl01 with a pre-merge upload of the nodepool image from Ibff32bb44e05413391dd7a320ba356f521bb30e8 once deployed.","accounts_in_message":[],"_revision_number":2},{"id":"03211e10003800545c5df3a8137fd2613fe37efe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:00:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e658bb5b539a751e7ea1c9419b23f66e3e4554c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:26:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b7f60e476d48b98f89ee199f881f39\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5d4519bd7a0d4fa4af444d67a78b30e7 : SUCCESS in 6m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/7f6ef38f656a49a083efc2ba341e4572 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b83dfd6ebac27d9c16c393487ef464be72cb31f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:26:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3074684e2b24947e3c14751c09b166e93f1cb8fe","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85690491bae4b108e18361fb107d277\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/9cdc42ebc22c4819a12e90e631727821 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eca3bde9cbba1b680f4f813a421ceb2d5803cf96","revisions":{"572b96d5cce52201937f3a2c86dc43fc79cbc64b":{"kind":"REWORK","_number":1,"created":"2024-03-05 22:19:05.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/81/911381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/911381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/911381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/911381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/911381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/911381/1"}}},"commit":{"parents":[{"commit":"56c5fefcf6394d6ebffd7407d990215a7973edc0","subject":"CentOS 7 removal prep changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56c5fefcf6394d6ebffd7407d990215a7973edc0"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-05 22:16:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-05 22:16:42.000000000","tz":0},"subject":"Try switching Rackspace DFW to an API key","message":"Try switching Rackspace DFW to an API key\n\nSwitch the Rackspace region with the smallest quota to uploading\nimages and booting server instances with our account\u0027s API key\ninstead of its password, in preparation for their MFA transition. If\nthis works as expected, we\u0027ll make a similar switch for the\nremaining two regions.\n\nChange-Id: I97887063c735c96d200ce2cbd8950bbec0ef7240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/572b96d5cce52201937f3a2c86dc43fc79cbc64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/572b96d5cce52201937f3a2c86dc43fc79cbc64b"}]},"branch":"refs/heads/master"},"eca3bde9cbba1b680f4f813a421ceb2d5803cf96":{"kind":"REWORK","_number":2,"created":"2024-03-06 15:06:43.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/81/911381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/911381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/911381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/911381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/911381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/911381/2"}}},"commit":{"parents":[{"commit":"56c5fefcf6394d6ebffd7407d990215a7973edc0","subject":"CentOS 7 removal prep changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56c5fefcf6394d6ebffd7407d990215a7973edc0"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-05 22:16:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-06 15:06:34.000000000","tz":0},"subject":"Try switching Rackspace DFW to an API key","message":"Try switching Rackspace DFW to an API key\n\nSwitch the Rackspace region with the smallest quota to uploading\nimages and booting server instances with our account\u0027s API key\ninstead of its password, in preparation for their MFA transition. If\nthis works as expected, we\u0027ll make a similar switch for the\nremaining two regions.\n\nChange-Id: I97887063c735c96d200ce2cbd8950bbec0ef7240\nDepends-On: https://review.opendev.org/911164\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eca3bde9cbba1b680f4f813a421ceb2d5803cf96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eca3bde9cbba1b680f4f813a421ceb2d5803cf96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
