)]}'
{"id":"openstack%2Fproject-config~911946","triplet_id":"openstack%2Fproject-config~master~Ib97bdda5d4e86714b674f560bec6bb7f288ff4e1","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"28439":{"account":{"_account_id":28439,"name":"Matt Peters","email":"matt.peters@windriver.com","username":"mpeters-wrs"},"last_update":"2024-03-29 00:00:17.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2024-03-28 18:00:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"28459":{"account":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},"last_update":"2024-03-19 14:31:32.000000000","reason":"\u003cGERRIT_ACCOUNT_28459\u003e replied on the change","reason_account":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-03-19 15:11:00.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib97bdda5d4e86714b674f560bec6bb7f288ff4e1","subject":"Add Distributed Cloud App to StarlingX","status":"MERGED","created":"2024-03-07 16:52:32.000000000","updated":"2024-03-29 01:12:06.000000000","submitted":"2024-03-29 00:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911946-new-project","meta_rev_id":"e5422e6b023f933eebf8011a93c697cde9e9257d","_number":911946,"virtual_id_number":911946,"owner":{"_account_id":28439,"name":"Matt Peters","email":"matt.peters@windriver.com","username":"mpeters-wrs"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-29 00:00:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},"all":[{"value":1,"date":"2024-03-28 18:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":1,"date":"2024-03-19 14:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-19 15:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-28 23:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 17:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-19 14:04:27.000000000","updated_by":{"_account_id":28439,"name":"Matt Peters","email":"matt.peters@windriver.com","username":"mpeters-wrs"},"reviewer":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},"state":"REVIEWER"},{"updated":"2024-03-19 14:31:32.000000000","updated_by":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-03-19 14:31:32.000000000","updated_by":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"}],"messages":[{"id":"9970921dd1528d9d35153739d8f4c52569716892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28439,"name":"Matt Peters","email":"matt.peters@windriver.com","username":"mpeters-wrs"},"date":"2024-03-07 16:52:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"414fd8b76881453ab8e176ab26c2efbc4c3fa666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 17:19:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0987e4fa68284800abaf76b98e1275f4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d036f8ab04b49fa9b3db81f0d17200f : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/518c890a13a144f79ec626e6788f8186 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"396b8a1d294b74365f34c5cfc7ac44a648ab5457","author":{"_account_id":28459,"name":"Bob Church","email":"robert.church@windriver.com","username":"rchurch"},"date":"2024-03-19 14:31:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dcc0a2d6f104e25faa0027658b0d976b10c85336","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-19 15:11:00.000000000","message":"Patch Set 1: Code-Review+2\n\nHolding approval so Scott can confirm this meets with the StarlingX community\u0027s expectations.","accounts_in_message":[],"_revision_number":1},{"id":"9f04db894db8cb0890657c2a6e746112d9b1fdc6","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2024-03-28 18:00:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"966611788732aa6a885d167e0fb86f2f4ade3eb7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-28 23:53:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"506d481494341140e3f79e50f483fee2a40a658d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:53:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64f110307b628f4d9ed5c25f020f1791b18f88e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 00:00:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0af128f91a8743b1b7d8cf530ec2ee7f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3eacd579d4414ccf96d600f4812bec34 : SUCCESS in 6m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ca343b67684f46f090558b6d099a9c4f : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d5498dd35caa05888917f55b9a73e4ba5160b150","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 00:00:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18b3c37bce645ef292b5cbc171cf8e6c947508fc","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 00:10:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ce3875bebd4cb79cc774e25123271c\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/90d2a5127ea24a998a0b43502c208daf : SUCCESS in 3m 42s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/bab74b05ae1a4802878ae7d2238de83b : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e5422e6b023f933eebf8011a93c697cde9e9257d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-29 01:12:06.000000000","message":"Patch Set 1:\n\nI\u0027ve added Scott as the initial member of the new starlingx-app-distributed-cloud-core group just now.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63e27662bbed22f0f03ba3f7f4a2e577f863d0ef","revisions":{"63e27662bbed22f0f03ba3f7f4a2e577f863d0ef":{"kind":"REWORK","_number":1,"created":"2024-03-07 16:52:32.000000000","uploader":{"_account_id":28439,"name":"Matt Peters","email":"matt.peters@windriver.com","username":"mpeters-wrs"},"ref":"refs/changes/46/911946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/46/911946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/911946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/911946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/911946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/46/911946/1"}}},"commit":{"parents":[{"commit":"3dccf06946bca6fbed4c1d58f72a0502e2f1c956","subject":"Merge \"Try switching Rackspace DFW to an API key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dccf06946bca6fbed4c1d58f72a0502e2f1c956"}]}],"author":{"name":"Matt Peters","email":"matt.peters@windriver.com","date":"2024-03-07 16:46:51.000000000","tz":-300},"committer":{"name":"Matt Peters","email":"matt.peters@windriver.com","date":"2024-03-07 16:52:19.000000000","tz":-300},"subject":"Add Distributed Cloud App to StarlingX","message":"Add Distributed Cloud App to StarlingX\n\nIntroduce the Distributed Cloud FluxCD application\nto StarlingX.\n\nChange-Id: Ib97bdda5d4e86714b674f560bec6bb7f288ff4e1\nSigned-off-by: Matt Peters \u003cmatt.peters@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/63e27662bbed22f0f03ba3f7f4a2e577f863d0ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/63e27662bbed22f0f03ba3f7f4a2e577f863d0ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
