)]}'
{"id":"openstack%2Fproject-config~912304","triplet_id":"openstack%2Fproject-config~master~I838f2ca24b429b620d3e286d634b85918fc40d6d","project":"openstack/project-config","branch":"master","topic":"drop-centos7","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-03-08 20:05:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I838f2ca24b429b620d3e286d634b85918fc40d6d","subject":"Drop gating for x/monitorstack","status":"MERGED","created":"2024-03-08 19:47:19.000000000","updated":"2024-03-08 20:26:46.000000000","submitted":"2024-03-08 20:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912304-drop-centos7","meta_rev_id":"a7afc04177bb59ee9432e8283dd5838e9e7e9a5d","_number":912304,"virtual_id_number":912304,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 20:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-03-08 19:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2024-03-08 19:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-08 19:51:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-03-08 19:53:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0ceaf4c2b30ec4c92a4bf3aa9bbb005ab501de1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-08 19:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f012084f9423e3f94e375a0468827fc86e6facbe","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-08 19:51:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da6f64e4e1c879e1e02628a06304e48f520a613d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:53:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a882f9e8bad94ca2a58dac8f63a13d95\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7cd21d9072284550812163e19086c5f8 : SUCCESS in 5m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/307a4cbf6dfc4a8cb3632d3eb13eca11 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bcda697e064971081c6b739cc42f2547b14a62af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:54:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e15851280d2eb2305cc47dd0d6318a604cc76ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 20:05:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7454c4ee3c347f8bd41f4a1004d43bf\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5f69f5229cf440aa9d258c8c0997adec : SUCCESS in 6m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5c0a48df1e7f4cfe8c63d171e701c4c3 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"416ad46f711952c81131c4076d6c304a7a4ffa7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 20:05:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7afc04177bb59ee9432e8283dd5838e9e7e9a5d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 20:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad0ab79c17774746a39b1fb4e26ed93d\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/05c63b39130c4809b83f9a9b1e39b777 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcea08868a8bafdc5f54bc5d440d1fefc76ca527","revisions":{"fcea08868a8bafdc5f54bc5d440d1fefc76ca527":{"kind":"REWORK","_number":1,"created":"2024-03-08 19:47:19.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/04/912304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/912304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/912304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/912304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/912304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/912304/1"}}},"commit":{"parents":[{"commit":"f315820135eafa9027c77b9a22af04d4c23d82e2","subject":"Drop CentOS 7 wheel builds and bindep fallback check job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f315820135eafa9027c77b9a22af04d4c23d82e2"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-08 19:44:38.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-08 19:44:38.000000000","tz":0},"subject":"Drop gating for x/monitorstack","message":"Drop gating for x/monitorstack\n\nRemove x/monitorstack from zuul. This repo uses CentOS 7 roles that\nwe want to clean up and it last merged a change in 2019 prior to\nbeing moved out of the openstack/ Git namespace. If interest\nrestarts it can be readded and fixed up.\n\nChange-Id: I838f2ca24b429b620d3e286d634b85918fc40d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcea08868a8bafdc5f54bc5d440d1fefc76ca527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcea08868a8bafdc5f54bc5d440d1fefc76ca527"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
