)]}'
{"id":"openstack%2Fproject-config~916007","triplet_id":"openstack%2Fproject-config~master~I1ec3e19c0dd8c83382e803525dd0b9612cfa75f8","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2024-04-17 12:29:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ec3e19c0dd8c83382e803525dd0b9612cfa75f8","subject":"Add openstack-helm-plugin git repository","status":"MERGED","created":"2024-04-16 17:07:55.000000000","updated":"2024-04-17 12:42:32.000000000","submitted":"2024-04-17 12:29:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916007","meta_rev_id":"f48c87e98fd9b8761ae43a9cc2108adebfcdd3ff","_number":916007,"virtual_id_number":916007,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 12:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-17 12:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-17 12:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 17:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 12:14:30.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"941ad761a0a99454dc329dccb8780513da773e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-16 17:07:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8211a3591bdb21f1fee0c61c193ef44a0783e823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 17:08:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/openstack-helm-plugin\" was not found.  All\n  projects referenced within a Zuul configuration must first be added to\n  the main configuration file by the Zuul administrator.\n\nThe problem appears in the the \"openstack/openstack-helm-plugin\" project stanza:\n\n  project:\n      name: openstack/openstack-helm-plugin\n      templates:\n        - official-openstack-repo-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3188","accounts_in_message":[],"_revision_number":1},{"id":"940cc16565182919be795f6793c3ac69a9d51df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-16 17:22:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eac3ed210c9c5f094beb0f2fdb8db4109dcd5440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 17:33:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca59ed76e41a4798ad1895ac00faf1fc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/102ca980bc0e479ab17be1fb07c7b3ad : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0757b4236b3b4647929abd9eab960b5f : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a70848bd7db28751ada2b905bc98723fcbb5065c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-16 17:37:52.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e688f76bd6e002abe16c9ad9c5d244b8022d1a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 17:49:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb76588fdc94a068b4fdbf3c58f0e4e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a7bc23d2e524590b7da9adff1f8fb44 : SUCCESS in 6m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/62d3a079ab47409e8b4c4f45ce82e769 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6a01d45ae253d59ed997783ced522406b2bb8dc7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-17 12:14:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nTaking the PTL\u0027s ownership of this change as implicit acknowledgement.","accounts_in_message":[],"_revision_number":3},{"id":"202c447d6ce6f64b35210c2b23030d80a83b2bae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:15:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ed6684dae39c937aeec4c4519a06872c4915159","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:29:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d21357d92b3e4fe38d3e656a7c6bfcf9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2ac874e1ac26464487e306e3a204f7d8 : SUCCESS in 7m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/acbe28c8a1f74556a860170df44a7fba : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8372904cd07867fb64408ba3520546732cfcff7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:29:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f48c87e98fd9b8761ae43a9cc2108adebfcdd3ff","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:42:32.000000000","message":"Patch Set 3:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41dc590e1aa242bfbff9f647af4ab37d\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/99fd9fa351bf431fb9443152b0bf5fec : FAILURE in 3m 47s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/783a1213643f4078887e65a64355185a : SKIPPED Skipped due to failed job infra-prod-manage-projects","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3dc66f157af58c2b2e48bd28349398e3eed9cd69","revisions":{"d6a97d242db5d327c96bcae94a14436e6f577c16":{"kind":"REWORK","_number":1,"created":"2024-04-16 17:07:55.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/07/916007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/916007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/916007/1"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-16 16:57:25.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-16 16:58:03.000000000","tz":-300},"subject":"Add openstack-helm-plugin git repository","message":"Add openstack-helm-plugin git repository\n\nThis repository will contain the code of the\nOSH Helm plugin which provides some convenience\nfor end users while deploying Openstack using\nOSH Helm charts.\n\nHelm allows installing plugins from git repos but\nit assumes cloning the whole git repo to the user\u0027s machine.\nTo avoid this we create this separate repository which\nwill contain only the plugin code.\n\nChange-Id: I1ec3e19c0dd8c83382e803525dd0b9612cfa75f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6a97d242db5d327c96bcae94a14436e6f577c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6a97d242db5d327c96bcae94a14436e6f577c16"}]},"branch":"refs/heads/master"},"0138c408adfd68d7ae7f16e6605c10f0b2b9ace2":{"kind":"REWORK","_number":2,"created":"2024-04-16 17:22:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/07/916007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/916007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/916007/2"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-16 16:57:25.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-16 17:21:54.000000000","tz":-300},"subject":"Add openstack-helm-plugin git repository","message":"Add openstack-helm-plugin git repository\n\nThis repository will contain the code of the\nOSH Helm plugin which provides some convenience\nfor end users while deploying Openstack using\nOSH Helm charts.\n\nHelm allows installing plugins from git repos but\nit assumes cloning the whole git repo to the user\u0027s machine.\nTo avoid this we create this separate repository which\nwill contain only the plugin code.\n\nChange-Id: I1ec3e19c0dd8c83382e803525dd0b9612cfa75f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0138c408adfd68d7ae7f16e6605c10f0b2b9ace2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0138c408adfd68d7ae7f16e6605c10f0b2b9ace2"}]},"branch":"refs/heads/master"},"3dc66f157af58c2b2e48bd28349398e3eed9cd69":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-04-16 17:37:52.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/07/916007/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/916007/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/916007/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/916007/3"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-16 16:57:25.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-16 17:36:20.000000000","tz":-300},"subject":"Add openstack-helm-plugin git repository","message":"Add openstack-helm-plugin git repository\n\nThis repository will contain the code of the\nOSH Helm plugin which provides some convenience\nfor end users while deploying Openstack using\nOSH Helm charts.\n\nHelm allows installing plugins from git repos but\nit assumes cloning the whole git repo to the user\u0027s machine.\nTo avoid this we create this separate repository which\nwill contain only the plugin code.\n\nNeeded-By: Ib4b5e3baa22f87a19b3f75952eb8b587acd8283b\nChange-Id: I1ec3e19c0dd8c83382e803525dd0b9612cfa75f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dc66f157af58c2b2e48bd28349398e3eed9cd69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dc66f157af58c2b2e48bd28349398e3eed9cd69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
