)]}'
{"id":"openstack%2Fproject-config~916038","triplet_id":"openstack%2Fproject-config~master~I00ced43f4bdadb0b230a6650f04cf20edab54964","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2024-04-23 13:32:43.000000000","reason":"removed on reply"},"36964":{"account":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"last_update":"2024-04-23 13:43:47.000000000","reason":"Change was submitted"},"29671":{"account":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"last_update":"2024-04-23 13:43:47.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-04-19 11:36:10.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I00ced43f4bdadb0b230a6650f04cf20edab54964","subject":"Add new repository powertrain-build","status":"MERGED","created":"2024-04-17 07:06:20.000000000","updated":"2024-04-23 13:53:32.000000000","submitted":"2024-04-23 13:43:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916038-new-project","meta_rev_id":"1fc4728715b197e034ecbf857d498f76688e17a9","_number":916038,"virtual_id_number":916038,"owner":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-23 13:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2024-04-23 13:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-22 12:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2024-04-23 13:32:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 07:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 08:16:03.000000000","updated_by":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2024-04-17 12:20:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-04-17 13:24:18.000000000","updated_by":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"reviewer":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"state":"CC"},{"updated":"2024-04-19 11:23:29.000000000","updated_by":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"reviewer":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"state":"REVIEWER"}],"messages":[{"id":"439e50faad3df0d481b8f1a78b99281b4d60eccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-17 07:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"601d1cc5ef9e55ffcffb45246dafb4402acd6f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 07:18:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6127e045261a40b0a32962d927ecc127\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4acd6298cfcc4d95aef5b1cf2a4fa0fb : SUCCESS in 6m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b5849df7031f4e3cb67559eb42314f27 : FAILURE in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"28bcab3eecefbb6e238ffe98a0827992504335f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-17 07:29:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0dc5c09eb3759ee8de9375abb4b388c3daaee38e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-17 07:31:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2dc409f910e12456ae45f4f3bd14e42392eada2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 07:40:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abec2e54aee64dbb919c933f3ab0485f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6287fc79aac24a56b170cb2df262bed3 : SUCCESS in 7m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/be046b4e6b8a4fcc8fd9d0f48654263b : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8057e149fd6bd03bbc34d7104fe05fa131b53e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-17 07:43:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2ae7719766c408d8b3cf5badc12215932fa93631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:01:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27fc147ba8824f53a638926a18337b52\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/27e12526263b4f82aab20379833b101d : SUCCESS in 6m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0075a31e8ce54088b8df1d76b15bfe0b : FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c72a01a84aa94d35f77299ca334535d46cf7fa92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-17 08:05:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"529509e3ef632290e3308e7fe8767d8461566efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:12:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4439c0bd24ec47078f31208ed004042e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/06bdf0f81514454687d915c2d683f0dd : SUCCESS in 6m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/495c7378f2064587960bfde9e08e97a3 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":5},{"id":"e74b1f917df6c30a523c9d9603edb8044c86fefb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-17 12:20:17.000000000","message":"Patch Set 5: Code-Review+2\n\nI\u0027ll refrain from single-core approving this since it will create a brand new Git namespace (so additional eyes on it would be good), and because it appears an explicit review from James has been requested.\n\nIn particular, it would be good to double-check that you really intend to gate this in the shared \"openstack\" Zuul tenant rather than our more generic \"opendev\" Zuul tenant or possibly in a completely separate \"volvocars\" Zuul tenant.","accounts_in_message":[],"_revision_number":5},{"id":"9c6d529ef437bef63e364c97f856cc56ea7116e0","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2024-04-17 13:24:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d93fd5560133ab359f7e506ad86188e67b99f009","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-04-17 13:39:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4d237f2772283664153778ef297788496b3516ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 07:12:56.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2bcdb24a678fdcc1f77eeaff80d01de92a54c3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 07:26:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/569bc274e3f2451b846bd54b9df2bca5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7ced5c0965de4a7387b57020c8c99788 : SUCCESS in 4m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c21326e1925244e986c4f91117aa011e : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":6},{"id":"c2d328b19a8f650aaf09b250a55f5922d0cfdad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 07:30:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1e7a61889325f0029a82176a091be520a9095e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 07:39:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cee3cd54f1274b9eaf265496a6339ce4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae25e2bfedbe4ede8d558fa358e9c8e4 : SUCCESS in 6m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cf3ec629e6044983aa9eda39fa37e042 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":7},{"id":"3e1b6d204fba58c2b4da6a39683ef7c84795a5f4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-19 11:03:00.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"854047e147aa0b9e6be7f9160766bbd2d2318e3e","author":{"_account_id":29671,"name":"Albin Vass","email":"opendev@albinvass.com","username":"albin_vass"},"date":"2024-04-19 11:23:29.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"392feb03bfbabeb3c05b60f7c1d7230ed1066761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 11:27:27.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"21b61e969c928cd3ae19141451ecd80d99dd3b4f","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 11:28:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"edaeef087046babd2e26db7e445eee7ab4ab5b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 11:29:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"da57625e8891a5a777d834770ae1c6eece2e40de","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 11:29:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cbf2da16328ebb7e697b646f6ddde88cfcfd7c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 11:32:03.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"3afa991c490b86a25824dbe4f8d080ed4e149ac5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-19 11:36:51.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"45e2fde5b89efdadbbfd80ac36740a31abdd7011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-19 11:47:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0a0fe540f8dc8417ced9bed5da7aa15169869c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 11:56:17.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c411b6df5b914cbd92f0ccbe24c55031\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/997447acb2ce42cb9ec431d9b2e46e01 : SUCCESS in 5m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2370e3ddf627409da1ae2a4bd9e31595 : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":11},{"id":"f640d3bfed68fa2ae2a7f1142ffec9be42e988d3","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-04-19 14:50:04.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f3fe3e7bae566c42b54d3d57d123dc5f7a1a4067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-22 06:49:28.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"754bd95648262498bfe1ce8cff3daa67b22a715b","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-22 06:55:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6161b13415f8921e9f865ac507f3e5cc45a5de84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-22 06:57:46.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d5ead89a82ba972d4274811fcd036dfd6a5f8c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 07:06:38.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4ed24890504ba68668f4bbd7b9bba6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5facd152edff41408cdddd3ac11d582d : SUCCESS in 7m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/83366036f94542a196600b30367478b5 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":13},{"id":"913e48bd1d9d55752fea2a3dfbfbc9ae9ae03ff3","author":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"date":"2024-04-22 07:07:46.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9bb85abd69df2641830a2a44bd43d08e5db7d3e9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-22 12:45:57.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"84f1d43e2131e5fb0cdd340214bc3a6b5d833591","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-04-23 13:32:43.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"c7c698c85fd30a4c6c5d0b652b4a0f602b99cd46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:33:02.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"9a380b8ae50144069d2016641989585c0ca66e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:43:47.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5552fd4c347d45d2951eed80327aaad2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/888c52f0686d4a969e188c1785d83340 : SUCCESS in 5m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c3954842d9c64e8088646271d2a5b0a3 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":13},{"id":"d035e6889a834c5b4e9b483b5bfe327eb8ccb8af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:43:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"1fc4728715b197e034ecbf857d498f76688e17a9","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:53:32.000000000","message":"Patch Set 13:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c1c8e4d92984179b6c7ceb3c1c8daae\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0aebf29c13dd4c1f8e5b960517afce01 : SUCCESS in 3m 52s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/cd830fb470c04119b1b3b7b844d35d42 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"b67a48c1f370976a2d42bdfea94e636c5724584b","revisions":{"2f819aba0e777491589b61097eacc664d57c96f4":{"kind":"REWORK","_number":1,"created":"2024-04-17 07:06:20.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/1"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f819aba0e777491589b61097eacc664d57c96f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f819aba0e777491589b61097eacc664d57c96f4"}]},"branch":"refs/heads/master"},"b967e45106d2d51ab8a3e124cde6950646d3802a":{"kind":"REWORK","_number":2,"created":"2024-04-17 07:29:29.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/2"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:29:09.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b967e45106d2d51ab8a3e124cde6950646d3802a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b967e45106d2d51ab8a3e124cde6950646d3802a"}]},"branch":"refs/heads/master"},"808080a91a814ac8a4896bd43888792c565bb05d":{"kind":"REWORK","_number":3,"created":"2024-04-17 07:31:45.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/3"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:31:34.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/808080a91a814ac8a4896bd43888792c565bb05d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/808080a91a814ac8a4896bd43888792c565bb05d"}]},"branch":"refs/heads/master"},"950b478552f42bd75815496f6136e3523103bc02":{"kind":"REWORK","_number":4,"created":"2024-04-17 07:43:44.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/4"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:43:31.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/950b478552f42bd75815496f6136e3523103bc02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/950b478552f42bd75815496f6136e3523103bc02"}]},"branch":"refs/heads/master"},"4fd6afe412931bbb50a4df334d8fb238079a3c5e":{"kind":"REWORK","_number":5,"created":"2024-04-17 08:05:02.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/5"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 08:04:54.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4fd6afe412931bbb50a4df334d8fb238079a3c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4fd6afe412931bbb50a4df334d8fb238079a3c5e"}]},"branch":"refs/heads/master"},"f4f90c0744a0476b76e7d540e3f3236aea0a9668":{"kind":"REWORK","_number":6,"created":"2024-04-19 07:12:56.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/6"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-19 07:11:57.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4f90c0744a0476b76e7d540e3f3236aea0a9668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f4f90c0744a0476b76e7d540e3f3236aea0a9668"}]},"branch":"refs/heads/master"},"724a1922e235b609f5296b0616bbe0f7404ffc6e":{"kind":"REWORK","_number":7,"created":"2024-04-19 07:30:57.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/7"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-19 07:30:48.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/724a1922e235b609f5296b0616bbe0f7404ffc6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/724a1922e235b609f5296b0616bbe0f7404ffc6e"}]},"branch":"refs/heads/master"},"0f6c5fe16dc645d6cc0714c20aa6e2f00fd0fbfa":{"kind":"REWORK","_number":8,"created":"2024-04-19 11:27:27.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/8"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-19 11:26:38.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\nand a new project-config repo was requested for handlding\nconfiguration for the \"volvocars\" namespace.\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f6c5fe16dc645d6cc0714c20aa6e2f00fd0fbfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f6c5fe16dc645d6cc0714c20aa6e2f00fd0fbfa"}]},"branch":"refs/heads/master"},"5f7ab0a5836c0cc5ca1f13bec9f196a7246b0d89":{"kind":"REWORK","_number":9,"created":"2024-04-19 11:29:36.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/9"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-19 11:29:28.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\nand a new project-config repo was requested for handlding\nconfiguration for the \"volvocars\" namespace.\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f7ab0a5836c0cc5ca1f13bec9f196a7246b0d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f7ab0a5836c0cc5ca1f13bec9f196a7246b0d89"}]},"branch":"refs/heads/master"},"543522201fc7157cdafacca6399c54a368384b7a":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2024-04-19 11:32:03.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/10"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-19 11:31:40.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\nand a new project-config repo was requested for handling\nconfiguration for the \"volvocars\" namespace.\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/543522201fc7157cdafacca6399c54a368384b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/543522201fc7157cdafacca6399c54a368384b7a"}]},"branch":"refs/heads/master"},"b187f6beb08181a887c7f72c0f55e7b549a56983":{"kind":"REWORK","_number":11,"created":"2024-04-19 11:47:00.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/11"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-19 11:46:52.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\nand a new project-config repo was requested for handling\nconfiguration for the \"volvocars\" namespace.\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b187f6beb08181a887c7f72c0f55e7b549a56983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b187f6beb08181a887c7f72c0f55e7b549a56983"}]},"branch":"refs/heads/master"},"94cc8ad2df307d0553119f3369b1ac655e15ad59":{"kind":"REWORK","_number":12,"created":"2024-04-22 06:49:28.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/12"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-22 06:49:16.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\nand a new project-config repo was requested for handling\nconfiguration for the \"volvocars\" namespace.\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94cc8ad2df307d0553119f3369b1ac655e15ad59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94cc8ad2df307d0553119f3369b1ac655e15ad59"}]},"branch":"refs/heads/master"},"b67a48c1f370976a2d42bdfea94e636c5724584b":{"kind":"REWORK","_number":13,"created":"2024-04-22 06:57:46.000000000","uploader":{"_account_id":36964,"name":"Alberto Gonzalez","display_name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","username":"albexl"},"ref":"refs/changes/38/916038/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/916038/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/916038/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/916038/13"}}},"commit":{"parents":[{"commit":"e14c4fbf962f0d98c066c3435943317cf1bdc446","subject":"Merge \"Retire puppet-sahara: Remove Project from Infrastructure System\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e14c4fbf962f0d98c066c3435943317cf1bdc446"}]}],"author":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-17 07:04:30.000000000","tz":120},"committer":{"name":"Alberto Gonzalez","email":"alberto.gonzalez@volvocars.com","date":"2024-04-22 06:57:38.000000000","tz":120},"subject":"Add new repository powertrain-build","message":"Add new repository powertrain-build\n\nThe repository was added to a new tenant named \"volvocars\"\nand a new project-config repo was requested for handling\nconfiguration for the \"volvocars\" namespace.\n\nChange-Id: I00ced43f4bdadb0b230a6650f04cf20edab54964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b67a48c1f370976a2d42bdfea94e636c5724584b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b67a48c1f370976a2d42bdfea94e636c5724584b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
