)]}'
{"id":"openstack%2Fproject-config~916876","triplet_id":"openstack%2Fproject-config~master~I98a57f9952392038acba73d9d5bfc5a9ecf81e7a","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"35904":{"account":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"last_update":"2024-05-03 22:08:47.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2024-05-03 22:08:47.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-04-29 13:03:36.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I98a57f9952392038acba73d9d5bfc5a9ecf81e7a","subject":"Add new repository for NebulOuS testing data","status":"MERGED","created":"2024-04-24 09:46:09.000000000","updated":"2024-05-03 22:27:32.000000000","submitted":"2024-05-03 22:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916876","meta_rev_id":"64fda92bc03c7e24734b6edd70e2acf955b59bf8","_number":916876,"virtual_id_number":916876,"owner":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 22:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2024-05-02 10:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-03 21:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-03 21:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 09:54:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-24 12:22:59.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2024-04-25 12:01:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-04-29 13:17:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2024-05-02 10:57:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"bb435d8d7fe5d6f01acb59d3d16774880b50b7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-04-24 09:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eea26b05aace61049b7e93c8529902ff4f016693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 09:54:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a475e62327a47cfa6ed875debb00ccf\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8a1f144048ff4f2db10db9e636cbb14a : SUCCESS in 6m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4aba4432f7664baba5ccc815c4f7d22d : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"19498decfd21ca2b52a240d9c0ae5ad0a8382b19","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-24 12:22:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecdf0b9b70eeef2ccb05322e96d996c42f742d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-04-25 08:13:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c655f806d0da378c7303964f55fc2e1c749c0225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 08:24:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/806dc1e11c034944aaee655af3cfdb17\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/478c21ad35034d679436654c6e63d132 : SUCCESS in 6m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/998569eadb1f497fb5a512fe55d85fd0 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f385ff1c51d11ad70cdf3127ee4bc693d02a6649","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-25 12:01:51.000000000","message":"Patch Set 2: Code-Review+2\n\nNot approving yet since Radosław typically provides confirmation for nebulous/ namespace additions.","accounts_in_message":[],"_revision_number":2},{"id":"53d87c69b4081593e0dc4a54173b0c5e969e2767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"date":"2024-04-29 13:00:09.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f5264a6484c3780372c2d00c1fdf6dc2f2f63a09","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-29 13:03:36.000000000","message":"Patch Set 3: Code-Review+2\n\nHolding approval until Radosław has a chance to confirm this addition is desired.","accounts_in_message":[],"_revision_number":3},{"id":"5854950760f2246f4e233b8e33670a8c9127f1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:11:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1e63e931ad4e379aeb8f11f9855767\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8b75ec56374348f79e3ec4afb81fa496 : SUCCESS in 5m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dc36c855006e4acebee07df28461c5b9 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"a1157fca903543353ac01bcf05f4712d244de9ae","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-04-29 13:17:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"518cee5b97782372b2323902ef09c35247bfb7df","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-04-29 14:15:51.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"beacdb28ed58cd55000fbc070f99671e20bb9fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-05-02 10:56:14.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8697b1322c65028d152c3265f03dab8f6c18e316","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-05-02 10:57:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7574c0afd6f6731d7c9da726774fa30f665d6b4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-05-02 10:57:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6642a6a2888f40f77d1d6c513eb01e336904414e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:08:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf50f7359c04b8b80ff282b7633a16b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/92f56faad240477a92373f8526a8fdd6 : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5d8ce93bf70849eebf45265a1f5c3d76 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"b49f88a0820ec792f42555ca3c37207a0275ee30","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-03 21:58:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5806bd9a9b37f6ab63984fe8414abfeb5dc91695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 21:59:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aa8c5e9599352e494867d9691917777f3651a126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 22:08:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bcd8c5d15e64500a73213a67f4a2994\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/907aa56178e9417a8423cead24b0e6b0 : SUCCESS in 7m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ae3ec50be5a94a96b000e411f81c8c5a : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":4},{"id":"fb4d4a312f03657fc8707f8fee8335bfca38d10c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 22:08:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"64fda92bc03c7e24734b6edd70e2acf955b59bf8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 22:27:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9b35b8c417448d0a6f4ab5ca243c59d\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/79e6f8d7a4594e2bb7e31ce8355c95a4 : SUCCESS in 3m 43s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/5347a875762b4d6f82712eac51b60ae2 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"783c235abb7a689aba196d38fd68c2c63f4f4477","revisions":{"1d3906fafc2e45b5bacb0e10008d2cf74d0518b9":{"kind":"REWORK","_number":1,"created":"2024-04-24 09:46:09.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/76/916876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/916876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/916876/1"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-04-24 09:46:01.000000000","tz":120},"committer":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-04-24 09:46:01.000000000","tz":120},"subject":"Add new repository for NebulOuS testing data","message":"Add new repository for NebulOuS testing data\n\nChange-Id: I98a57f9952392038acba73d9d5bfc5a9ecf81e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d3906fafc2e45b5bacb0e10008d2cf74d0518b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d3906fafc2e45b5bacb0e10008d2cf74d0518b9"}]},"branch":"refs/heads/master"},"4ffff6df434c9cf5402f95f68b8b5272f7371d34":{"kind":"REWORK","_number":2,"created":"2024-04-25 08:13:29.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/76/916876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/916876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/916876/2"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-04-24 09:46:01.000000000","tz":120},"committer":{"name":"Jan Marchel","email":"jmarchel@7bulls.com","date":"2024-04-25 08:13:29.000000000","tz":0},"subject":"Add new repository for NebulOuS testing data","message":"Add new repository for NebulOuS testing data\n\nChange-Id: I98a57f9952392038acba73d9d5bfc5a9ecf81e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ffff6df434c9cf5402f95f68b8b5272f7371d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ffff6df434c9cf5402f95f68b8b5272f7371d34"}]},"branch":"refs/heads/master"},"18dc3142f3e4ac89a6a02268cfca982c06bc435b":{"kind":"REWORK","_number":3,"created":"2024-04-29 13:00:09.000000000","uploader":{"_account_id":35904,"name":"Jan Marchel","display_name":"Jan Marchel","email":"jmarchel@7bulls.com","username":"janmarchel"},"ref":"refs/changes/76/916876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/916876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/916876/3"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-04-24 09:46:01.000000000","tz":120},"committer":{"name":"Jan Marchel","email":"jmarchel@7bulls.com","date":"2024-04-29 12:59:59.000000000","tz":0},"subject":"Add new repository for NebulOuS testing data","message":"Add new repository for NebulOuS testing data\n\nChange-Id: I98a57f9952392038acba73d9d5bfc5a9ecf81e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18dc3142f3e4ac89a6a02268cfca982c06bc435b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18dc3142f3e4ac89a6a02268cfca982c06bc435b"}]},"branch":"refs/heads/master"},"783c235abb7a689aba196d38fd68c2c63f4f4477":{"kind":"REWORK","_number":4,"created":"2024-05-02 10:56:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/76/916876/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/916876/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/916876/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/916876/4"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"jmarchel","email":"jmarchel@7bulls.com","date":"2024-04-24 09:46:01.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radek@piliszek.it","date":"2024-05-02 10:56:14.000000000","tz":0},"subject":"Add new repository for NebulOuS testing data","message":"Add new repository for NebulOuS testing data\n\nChange-Id: I98a57f9952392038acba73d9d5bfc5a9ecf81e7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/783c235abb7a689aba196d38fd68c2c63f4f4477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/783c235abb7a689aba196d38fd68c2c63f4f4477"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
