)]}'
{"id":"openstack%2Fproject-config~917827","triplet_id":"openstack%2Fproject-config~master~I678ded14ff2e26ca235ba53a67991a56fc1a54a7","project":"openstack/project-config","branch":"master","topic":"sunbeam-noop-jobs","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-08-09 16:56:11.000000000","reason":"Change was submitted"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-08-09 16:29:39.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-09 16:25:48.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2024-08-09 16:56:11.000000000","reason":"Change was submitted"},"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2024-08-09 16:56:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I678ded14ff2e26ca235ba53a67991a56fc1a54a7","subject":"Prep sunbeam single charm repos for retirement","status":"MERGED","created":"2024-05-01 05:03:16.000000000","updated":"2024-08-09 16:56:11.000000000","submitted":"2024-08-09 16:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917827-sunbeam-noop-jobs","meta_rev_id":"4a43b122af8c6e3839f29ff11529f6cc11c009fb","_number":917827,"virtual_id_number":917827,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-09 16:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-08-09 16:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2024-08-09 16:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"value":2,"date":"2024-08-09 16:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2024-08-09 16:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 05:08:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-03 20:20:50.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-05-16 19:37:30.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-08-09 07:47:44.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2024-08-09 08:10:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-08-09 10:56:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-09 16:31:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"84058ca720b34caa08288cd8c3cb6aaa3be5b26a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-05-01 05:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d4fbd94d96cd6d1b2c792bf509899b266f3d0a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 05:08:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54626d8c8fa2419ea245a84fa6b504c4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be49c1a84b1d40158eff33087cd10a48 : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"72bb4a1c2cf64ffe329161ed0d056cb3d4a57e19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-05-01 05:12:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e76be879a2b00dd0af3cea9254625e6f548d2fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 05:24:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b11218b10b34be0b0870c9c4ad1ff74\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d31c4c16781c43c78a497c1752a57acc : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"90a8659939283305a90cbf5aa66d8e577f81af66","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-03 20:20:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f447216927d13f4e208aecf82ed13d0da314c9c9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-16 19:37:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e43d74de78f4cfa3d13e63a515923c277b6190e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-06-04 09:11:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"56625a656b06783bcc1c825c56a0e45501e9e80b","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-06-04 09:12:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"064fc613d50096044464cdb7d0a01265109343c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 09:20:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fce5409fbc24227aec705cfc6c91ead\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e8784586966d45e6b8de2e6296c8ba3c : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"16d0566e303d8771bc909982c40e9a88e67e5e6d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-24 07:19:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e704cdc24360127a47d5cfb4dd1de26200170f2f","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-07-05 08:53:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76cd423f8cfa174cbfdac776f663f4f3cc8c7869","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2024-08-09 07:47:44.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b66d518c1aa4b8e9e3cf80f35dcd59eb9ff455b5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 08:10:58.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b2fa27e300c2c50feb5f65a70757a319fca5a1c0","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-09 09:38:28.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"598d84373bfd96bba23f6e66353cdc41f43c5171","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 10:56:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"701c5dc39f1906993020d63d6169f380b94326e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-09 12:22:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"94f1776da4ca68a36805d63ea9a7ed1a1597995c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 12:29:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04473b00498b45fb871633f5b670c5cd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/28426f56c0e041aa996758f99dc4ca04 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":4},{"id":"6eb2cd6f3854b706368041be9d0f9bc0cb36cd03","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-08-09 15:40:24.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ac55fff2ff8a4ebb3b330d4ab1d11381448bd53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-09 15:51:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9844e87477a135eb4cdc11ad6fae011613cf6f6f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-08-09 15:52:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9cd9aebc20972e3e2af63695ea28f1a47d8c12f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:03:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fdd09ad410f4f26b18d0326c4fb0d7e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7923a81ea232419fb318b0b15c2c71ca : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":5},{"id":"df85f34236912035f7998101e6d38cda492f2fda","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 16:25:48.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ce5ee2bba82c3219f619978bb5915d54e12b1094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-09 16:28:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"863ecbdb6b3e98d49d9b1bf47e9ac85156d0666b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-08-09 16:29:39.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2dd7a4d1fa6c0355964b31fd7146e0db63d0d741","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-08-09 16:31:17.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"34e429fdd22337bb2258f5007817178c0d5db449","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 16:32:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"863a8a661714cc3377db27347af6b1fae56b2b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:41:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b1ddb77b9547938da67f38825aa86c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ba59153d46dd467289d677ed54af65a8 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":6},{"id":"be8f7adef0178c0643bb7e396866fd532f2e1a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:41:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"58fa73168b4b5b838d8608e0cb27054ccc6ab750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:56:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efceca393f1d4fe38cc337a349da2009\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ecded1c16144732b6f695a8abb0dde8 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":6},{"id":"4a43b122af8c6e3839f29ff11529f6cc11c009fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:56:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cd1d98c3df2b812200a7df56c5df0cde5cebb844","revisions":{"3e0bd8f38a4bb6c561c6afcbc14f368df351b93e":{"kind":"REWORK","_number":1,"created":"2024-05-01 05:03:16.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/27/917827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/917827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/917827/1"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"subject":"Setup/update Sunbeam single charm repos to noop","message":"Setup/update Sunbeam single charm repos to noop\n\nAs the retirement process requires that a retired repository cannot\ncontain anything other than a .gitreview and README, setup noop\njobs for all Sunbeam repositories being retired so we can meet this\nrequirement.\n\nChange-Id: I678ded14ff2e26ca235ba53a67991a56fc1a54a7\nRequired-By: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e0bd8f38a4bb6c561c6afcbc14f368df351b93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e0bd8f38a4bb6c561c6afcbc14f368df351b93e"}]},"branch":"refs/heads/master"},"8a1e67cfdb73c06b05c039cf670d7ac4ad42a8ad":{"kind":"REWORK","_number":2,"created":"2024-05-01 05:12:16.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/27/917827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/917827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/917827/2"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:12:08.000000000","tz":60},"subject":"Setup/update Sunbeam single charm repos to noop","message":"Setup/update Sunbeam single charm repos to noop\n\nAs the retirement process requires that a retired repository cannot\ncontain anything other than a .gitreview and README, setup noop\njobs for all Sunbeam repositories being retired so we can meet this\nrequirement.\n\nChange-Id: I678ded14ff2e26ca235ba53a67991a56fc1a54a7\nRequired-By: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a1e67cfdb73c06b05c039cf670d7ac4ad42a8ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a1e67cfdb73c06b05c039cf670d7ac4ad42a8ad"}]},"branch":"refs/heads/master"},"ffbf05070f87ece475e6657448bc5039fc05c51e":{"kind":"REWORK","_number":3,"created":"2024-06-04 09:11:22.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/27/917827/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/917827/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/917827/3"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-06-04 09:11:18.000000000","tz":60},"subject":"Prep sunbeam single charm repos for retirement","message":"Prep sunbeam single charm repos for retirement\n\nEnsure that all sunbeam single charm repos are covered by central\nzuul configuration so that the repositories can be retired as part\nof the move to the sunbeam-charms mono-repo.\n\nThis allows the in-repo zuul configuration to be removed which\ncurrently blocks retirement.\n\nChange-Id: I678ded14ff2e26ca235ba53a67991a56fc1a54a7\nRequired-By: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ffbf05070f87ece475e6657448bc5039fc05c51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ffbf05070f87ece475e6657448bc5039fc05c51e"}]},"branch":"refs/heads/master"},"ec0a800fce29d0a97416f658418c0941ceeaebf3":{"kind":"REWORK","_number":4,"created":"2024-08-09 12:22:34.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/27/917827/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/917827/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/917827/4"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-09 12:22:27.000000000","tz":60},"subject":"Prep sunbeam single charm repos for retirement","message":"Prep sunbeam single charm repos for retirement\n\nEnsure that all sunbeam single charm repos are covered by central\nzuul configuration so that the repositories can be retired as part\nof the move to the sunbeam-charms mono-repo.\n\nThis allows the in-repo zuul configuration to be removed which\ncurrently blocks retirement.\n\nChange-Id: I678ded14ff2e26ca235ba53a67991a56fc1a54a7\nRequired-By: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec0a800fce29d0a97416f658418c0941ceeaebf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec0a800fce29d0a97416f658418c0941ceeaebf3"}]},"branch":"refs/heads/master"},"76fa22dc37f3d31d2cc12bbffa32df9e8b26ca87":{"kind":"REWORK","_number":5,"created":"2024-08-09 15:51:17.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/27/917827/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/917827/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/917827/5"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-09 15:51:00.000000000","tz":60},"subject":"Prep sunbeam single charm repos for retirement","message":"Prep sunbeam single charm repos for retirement\n\nEnsure that all sunbeam single charm repos are covered by central\nzuul configuration so that the repositories can be retired as part\nof the move to the sunbeam-charms mono-repo.\n\nThis allows the in-repo zuul configuration to be removed which\ncurrently blocks retirement.\n\nChange-Id: I678ded14ff2e26ca235ba53a67991a56fc1a54a7\nRequired-By: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76fa22dc37f3d31d2cc12bbffa32df9e8b26ca87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76fa22dc37f3d31d2cc12bbffa32df9e8b26ca87"}]},"branch":"refs/heads/master"},"cd1d98c3df2b812200a7df56c5df0cde5cebb844":{"kind":"REWORK","_number":6,"created":"2024-08-09 16:28:20.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/27/917827/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/917827/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/917827/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/917827/6"}}},"commit":{"parents":[{"commit":"f0cde8e841746c428b3613160c8eb5a8d7f7e4ca","subject":"Merge \"Add reference to the project-team-guide in the fail ci msg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0cde8e841746c428b3613160c8eb5a8d7f7e4ca"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-05-01 05:01:04.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-09 16:28:14.000000000","tz":60},"subject":"Prep sunbeam single charm repos for retirement","message":"Prep sunbeam single charm repos for retirement\n\nEnsure that all sunbeam single charm repos are covered by central\nzuul configuration so that the repositories can be retired as part\nof the move to the sunbeam-charms mono-repo.\n\nThis allows the in-repo zuul configuration to be removed which\ncurrently blocks retirement.\n\nChange-Id: I678ded14ff2e26ca235ba53a67991a56fc1a54a7\nRequired-By: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd1d98c3df2b812200a7df56c5df0cde5cebb844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd1d98c3df2b812200a7df56c5df0cde5cebb844"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
