)]}'
{"id":"openstack%2Fproject-config~919396","triplet_id":"openstack%2Fproject-config~master~If7ef4a6351c397a3190b53593054a73ac354fdc7","project":"openstack/project-config","branch":"master","topic":"retire-ec2api","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-05-14 01:10:30.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-05-13 22:50:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If7ef4a6351c397a3190b53593054a73ac354fdc7","subject":"Retire ec2-api: end gate and update acl","status":"MERGED","created":"2024-05-11 01:02:19.000000000","updated":"2024-05-14 01:37:53.000000000","submitted":"2024-05-14 01:32:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919396-retire-ec2api","meta_rev_id":"934c4c86fe25357e02b5aee4880f7dc811332420","_number":919396,"virtual_id_number":919396,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 01:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-05-13 22:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-13 22:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-13 22:50:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-11 01:09:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 19:20:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-05-13 19:36:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"b6827f5095d920f5d7231bcea8f4e1a50330d092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-11 01:02:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7ea578b5f7dd55248b1506c0a0af4fed577bc52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 01:09:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d302b34409d4405a94f2843006faafb5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ab79efe6e8b541f390a19a3c9c681c1d : SUCCESS in 4m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/39c2287c46694ce694fd1b02d7af980d : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"40d090256a5fda4d3c41e21b02eea93a8e0989d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-11 01:10:12.000000000","message":"Topic retire-ec2-api removed","accounts_in_message":[],"_revision_number":1},{"id":"b512651a1e4a3cf0989ae80151a7a42bd55585cd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-11 01:10:27.000000000","message":"Topic set to retire-ec2api","accounts_in_message":[],"_revision_number":1},{"id":"e9488fd254ba2322868e6fd9dd16f415fc211de9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-13 19:20:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce7d4f2aa9707dc0981a11657075845697e0ecb6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-13 19:20:52.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"5924acd2ae8308516c5becf038aa7d4e54f51a50","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-13 19:36:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d6d87624237ce9c10cdbf29644f4767bc15347d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 19:36:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"542b70487d2133856c350590242c6d7cb92e66bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 19:41:26.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/919376 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"24fbd7bad5e5944ef5db1d284eb8365facd1fbcd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-13 22:07:53.000000000","message":"Patch Set 1:\n\nrecheck now that the parent is back in the gate","accounts_in_message":[],"_revision_number":1},{"id":"90d4a378bc05f6a4beca3aa696d08e33069372b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 22:08:27.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/project-config for 919376,2","accounts_in_message":[],"_revision_number":1},{"id":"a4b64498a822c122b2299b06948d7a8286bc51e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-13 22:40:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"404e6915f4dfc6b3aca4205d47ed9a411a5b2b5b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-13 22:50:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c3a61533f3fdea08d447ac9c21efec72aca76785","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-14 01:10:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b971714c1392b2d5fedf67e10f3e22c27bc7406a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 01:21:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/781d940044c749fba5f1b78fb26b440f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c2b79d81f6024a3daac5b649f002ed0e : SUCCESS in 4m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/632fdf1a469b4bbc914340270d1f2f6b : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"24633ab736a8326b7826bcc6f4af3b29ed6306e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 01:21:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e539b0d2bc1f6924543990003d7bcfc29f8ffac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 01:32:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c5e61dfe0cb407da54d65182e92a3d3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cebce477a42a42c1af9197b437d14efd : SUCCESS in 5m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2a3c604fd5624bc9985156f12e603cda : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"948c7c8e7cfcc5e3eb89b512d02df4d7039853aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 01:32:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"934c4c86fe25357e02b5aee4880f7dc811332420","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 01:37:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3678a90cecd40f1a26d54f0c251de89\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/e9a28aeb532a4baa84974fe3ae3b0c2f : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f96e1dcb9128a4546bc5fc81f35a0e55240713ff","revisions":{"ae0c2d3bb6618b131c929ff7fe7fa3d27a8bfc2f":{"kind":"REWORK","_number":1,"created":"2024-05-11 01:02:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/96/919396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/919396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/919396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/919396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/919396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/919396/1"}}},"commit":{"parents":[{"commit":"4e8c19f39aab10c201c8cfa64e41692579d7b1d4","subject":"Retire Sahara: end gate and update acl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e8c19f39aab10c201c8cfa64e41692579d7b1d4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-11 01:01:08.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-11 01:01:08.000000000","tz":-420},"subject":"Retire ec2-api: end gate and update acl","message":"Retire ec2-api: end gate and update acl\n\nec2-api project is retiring[1] and this is the first step\nof removing the gate jobs from this repo and also update acl so that\nwe can proceed on the remaining cleanup.\n\n[1] https://review.opendev.org/c/openstack/governance/+/919394/1\n\nChange-Id: If7ef4a6351c397a3190b53593054a73ac354fdc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae0c2d3bb6618b131c929ff7fe7fa3d27a8bfc2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae0c2d3bb6618b131c929ff7fe7fa3d27a8bfc2f"}]},"branch":"refs/heads/master"},"f96e1dcb9128a4546bc5fc81f35a0e55240713ff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-13 22:40:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/96/919396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/919396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/919396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/919396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/919396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/919396/2"}}},"commit":{"parents":[{"commit":"86c3c171bff9a60233dc52d48eb90ad7c56409b2","subject":"Retire Sahara: end gate and update acl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/86c3c171bff9a60233dc52d48eb90ad7c56409b2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-11 01:01:08.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-05-13 22:40:19.000000000","tz":0},"subject":"Retire ec2-api: end gate and update acl","message":"Retire ec2-api: end gate and update acl\n\nec2-api project is retiring[1] and this is the first step\nof removing the gate jobs from this repo and also update acl so that\nwe can proceed on the remaining cleanup.\n\n[1] https://review.opendev.org/c/openstack/governance/+/919394/1\n\nChange-Id: If7ef4a6351c397a3190b53593054a73ac354fdc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f96e1dcb9128a4546bc5fc81f35a0e55240713ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f96e1dcb9128a4546bc5fc81f35a0e55240713ff"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
