)]}'
{"id":"openstack%2Fproject-config~919397","triplet_id":"openstack%2Fproject-config~master~I00c599b5a89db53674b34ab2ae1f5e784e59e779","project":"openstack/project-config","branch":"master","topic":"retire-ec2api","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-05-17 09:14:20.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-05-17 08:58:23.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-05-17 09:14:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00c599b5a89db53674b34ab2ae1f5e784e59e779","subject":"Retire ec2-api: remove project from infra","status":"MERGED","created":"2024-05-11 01:05:33.000000000","updated":"2024-05-17 09:41:44.000000000","submitted":"2024-05-17 09:14:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919397-retire-ec2api","meta_rev_id":"72916585c80ced2671174f45efee269c2e161c4c","_number":919397,"virtual_id_number":919397,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-17 09:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-05-16 23:34:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2024-05-17 08:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2024-05-17 08:58:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-11 01:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 13:59:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-05-16 19:43:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"5bad4b1d1030afac2f4759429f573a9738e2223b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-11 01:05:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd45f39d263213bf858de5001f7cb6557b2b045d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 01:20:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/687c3cf87378413c971b682ec6347aef\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3ee731460aca4a6db70f49af6a94be8d : SUCCESS in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/57513bc6f9ff4a87bb7f1c92c181b8c4 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d4f08bdb1d47e5a8b1594dc77624be120349857b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 08:34:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5a672fafc264ffb37f77005f6bbd34ce5dc0210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 08:57:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab87e76969cd422a8695587eda2318ed\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/329eeb824efd4a64a4acdff8cc690d14 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e5ab467a57ce4adf90d47cc252674cea : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2be503302361d866b061b5d9f5f5bd20248e9971","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-16 13:59:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bd27012b7ec1373b1ad49821cfe7ab66422cfc5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-16 17:00:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60ff53f209d6c2994d67cc13510e6ff38f21293f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-16 19:43:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2872f89a63d21c754c5360aa738a1812ed605085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-16 23:34:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"46c847fa5adad115b6594351278a36c1c8cbdbcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 23:41:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd0fe05f6d64529932c60de6b985f7d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/913d912e9c0c463fa41251282913d8fe : SUCCESS in 5m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ace629bea98d448ea5a529dab6bc27eb : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f230f339233e5017c8afd9d75fd6e1d72a05fb11","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-17 08:58:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8f7893f4ce5cf32890d24fc02b408021d747487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 09:03:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d624f63740a4952d25877a0f728117a26b683e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 09:14:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15f23fb8737541dc80f1805b3f95f081\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eabe11e8825f41d7b1a47709c9a3e713 : SUCCESS in 7m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9333b3218ff246eda3edc49c0ecd65a1 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"5012ca2c302ae531995f5eb51b418460ee28950d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 09:14:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"72916585c80ced2671174f45efee269c2e161c4c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 09:41:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd384afe0f948fa94cfab6a78b1cdf7\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f5d041934c94417a862afacc5672a5c6 : SUCCESS in 3m 41s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/27aeb86ff7f140a689b1676933598e6f : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a222415469107d5600e6857ea88a7ae7d26bb6d","revisions":{"4d0399eac51dc2110fa932c1c0f45df840a93b62":{"kind":"REWORK","_number":1,"created":"2024-05-11 01:05:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/97/919397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/919397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/919397/1"}}},"commit":{"parents":[{"commit":"ae0c2d3bb6618b131c929ff7fe7fa3d27a8bfc2f","subject":"Retire ec2-api: end gate and update acl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae0c2d3bb6618b131c929ff7fe7fa3d27a8bfc2f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-11 01:04:34.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-11 01:04:34.000000000","tz":-420},"subject":"Retire ec2-api: remove project from infra","message":"Retire ec2-api: remove project from infra\n\nec2-api project has been retired, this commit removes\nit from infra.\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/919394/1\nChange-Id: I00c599b5a89db53674b34ab2ae1f5e784e59e779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d0399eac51dc2110fa932c1c0f45df840a93b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d0399eac51dc2110fa932c1c0f45df840a93b62"}]},"branch":"refs/heads/master"},"cfe0d4b480d70d1ba1333baf0f1ac6d31ca5a5f0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-15 08:34:22.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/97/919397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/919397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/919397/2"}}},"commit":{"parents":[{"commit":"dd1e4e61a758d61066d3e8cbdec10607a3f8cdff","subject":"Retire Sahara: remove project from infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd1e4e61a758d61066d3e8cbdec10607a3f8cdff"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-11 01:04:34.000000000","tz":-420},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-15 08:33:03.000000000","tz":120},"subject":"Retire ec2-api: remove project from infra","message":"Retire ec2-api: remove project from infra\n\nec2-api project has been retired, this commit removes\nit from infra.\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/919394/1\nChange-Id: I00c599b5a89db53674b34ab2ae1f5e784e59e779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cfe0d4b480d70d1ba1333baf0f1ac6d31ca5a5f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cfe0d4b480d70d1ba1333baf0f1ac6d31ca5a5f0"}]},"branch":"refs/heads/master"},"6a222415469107d5600e6857ea88a7ae7d26bb6d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-16 23:34:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/97/919397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/97/919397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/97/919397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/97/919397/3"}}},"commit":{"parents":[{"commit":"a8a40342490fd79ee11f284608cb955b011dcf3a","subject":"Retire Senlin: remove project from infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8a40342490fd79ee11f284608cb955b011dcf3a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-16 23:34:02.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-16 23:34:02.000000000","tz":-420},"subject":"Retire ec2-api: remove project from infra","message":"Retire ec2-api: remove project from infra\n\nec2-api project has been retired, this commit removes\nit from infra.\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/919394/1\nChange-Id: I00c599b5a89db53674b34ab2ae1f5e784e59e779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a222415469107d5600e6857ea88a7ae7d26bb6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a222415469107d5600e6857ea88a7ae7d26bb6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
