)]}'
{"id":"openstack%2Fproject-config~919630","triplet_id":"openstack%2Fproject-config~master~I50ce4e5aa7001ba52bea78a65855278be68e61a5","project":"openstack/project-config","branch":"master","topic":"drop-ubuntu-xenial","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-05-28 21:25:22.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}},"hashtags":[],"change_id":"I50ce4e5aa7001ba52bea78a65855278be68e61a5","subject":"Retire devstack-gate","status":"MERGED","created":"2024-05-14 22:18:07.000000000","updated":"2024-05-28 22:13:07.000000000","submitted":"2024-05-28 21:52:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919630-drop-ubuntu-xenial","meta_rev_id":"91dc591048e79d5bb457e2a18c41145fa976b991","_number":919630,"virtual_id_number":919630,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 21:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-05-17 11:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-05-15 07:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-05-15 01:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-05-17 11:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 22:35:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 01:53:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-05-15 07:53:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-05-17 11:53:01.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"a1bb8248e2ae6ce5e4be0ae8e89f3039411061a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-14 22:18:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"163715c8c4dc2fdb13c01ec82fab72bbfb88931f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 22:35:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3f5c1977c1541cab99920ee7047811f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/41582063ec9845b0b391dcc9e0c11b62 : SUCCESS in 5m 23s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/1cc34d026c264e49b1c399299bd5f079 : SUCCESS in 3m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b2d56e7be2c74fd5be3baebbb4e447bc : SUCCESS in 3m 29s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/584ce470f21742778c58c0d6c35ad397 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"451597781d377b7b5602a249aec4b5200758351d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 01:53:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"492f0d81805996ef041a3ed23081f1b43b8ae70d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 07:53:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99a8379bf3393209866540b605555a0793d5f9e7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-17 11:53:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a516e1e5658ce957b9e056cfbecfeab1564e6d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 11:53:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5be5ce085274f6d9ed2b991278f3303\n\n\nWarning:\n  Change 919629 in project openstack/governance does not share a change queue with 919630 in project openstack/project-config","accounts_in_message":[],"_revision_number":1},{"id":"7b06ea22af90af69205ca54d8c5c746fc0ee53ed","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-28 21:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25d2a67a6cfe5b7f92439f236ae192ceb0d83585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 21:41:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c40ebd02625a4057a37ad3cc6108ff2d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9dc37ae295bf425789f4a94776169283 : SUCCESS in 5m 03s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6d8557a8f37c40f98801d9faca5f52f0 : SUCCESS in 2m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/704e57e1b73242058ea9b5ea4ab1af82 : SUCCESS in 3m 22s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/18560bbc08994d7c8c8c8582007c7561 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1b72e65060bbb04a92f22602b66c2db75545b79e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 21:41:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27323ead8755b1a268cdd0567370e911d8ab6a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 21:52:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524f05cdb819445d832c600987ff842c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6c833730debc40dda54c2e0d0719491c : SUCCESS in 4m 57s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6854e52da29c480a9ea4281dfa5d9b7a : SUCCESS in 2m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/35c291a46fc646dfabd355f20faf7904 : SUCCESS in 3m 05s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7f81b72db93c4070ab58c65049874aad : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a4215d5d23105da83ba17893eb28b85df795dc6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 21:52:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91dc591048e79d5bb457e2a18c41145fa976b991","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 22:13:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d73f7c05bd1342cb8f511144ce290b37\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/cf059c57354f42fdb3be7f1795e661c6 : SUCCESS in 3m 42s\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/5b0d8082f49e412fb62786ed55d7e0af : SUCCESS in 6m 40s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/58c75e0fea9e45ff8138d4d442eeb5ed : SUCCESS in 5m 23s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/04701bf12a52425a821164e3bb323714 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82e7504c36d09a665e66180dc4fba3f206d06493","revisions":{"82e7504c36d09a665e66180dc4fba3f206d06493":{"kind":"REWORK","_number":1,"created":"2024-05-14 22:18:07.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/30/919630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/919630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/919630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/919630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/919630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/919630/1"}}},"commit":{"parents":[{"commit":"e7525f8f5286269ff456611a3f482879662b6aa2","subject":"Add devstack-gate back to central config for retirement prep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e7525f8f5286269ff456611a3f482879662b6aa2"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-14 22:16:58.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-14 22:16:58.000000000","tz":-420},"subject":"Retire devstack-gate","message":"Retire devstack-gate\n\nThis removes devstack-gate from the CI system and updates ACLS for the\nrepo in Gerrit to the openstack retired repo config. There are also\ncleanups to nodepool elements and irc bots to remove references to\ndevstack-gate where we don\u0027t need them anymore.\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/919629\nChange-Id: I50ce4e5aa7001ba52bea78a65855278be68e61a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82e7504c36d09a665e66180dc4fba3f206d06493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82e7504c36d09a665e66180dc4fba3f206d06493"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
