)]}'
{"id":"openstack%2Fproject-config~920779","triplet_id":"openstack%2Fproject-config~master~I7cb21869bae42baed5027a9380f60762ab8944e0","project":"openstack/project-config","branch":"master","topic":"zuul-git-cache-ownership","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-05-29 22:08:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7cb21869bae42baed5027a9380f60762ab8944e0","subject":"Chown the /opt/git repo cache to zuul:zuul","status":"MERGED","created":"2024-05-29 21:38:41.000000000","updated":"2024-05-29 22:24:06.000000000","submitted":"2024-05-29 22:08:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920779-zuul-git-cache-ownership","meta_rev_id":"120a635bd83af1aa0c0059c3914882768a36959c","_number":920779,"virtual_id_number":920779,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-29 22:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2024-05-29 21:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2024-05-29 22:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2024-05-29 21:55:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-29 21:44:53.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2024-05-29 21:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-29 22:02:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"cd3b3785daa474a279f8a64a2f93090df704d31e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-29 21:38:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6f4cc96f1ca4804d5132b00f3e5ead35c15e6cf","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-05-29 21:44:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"999f57ac6576dc7c799856ba83457ef2f3da730f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:50:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f04d3bcc44b4c8d9a0e339d86fafeb0\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5bc187a4bcb04328b8c8f34072868b51 : SUCCESS in 6m 05s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/9fbf1944a0314ec4ba0a4f9efb6e80d5 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f326177f7d1dd631fc4e3367e99a2029b5e86e0e","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-05-29 21:55:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47782c1ddd7e626bf045d5133642bc31544fbd74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:56:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d081187b76dc5cf5f4995c1e3e4a5c1d99471b17","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-29 22:02:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db899e81c9e079059f7d1e952022f32a659bdfd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 22:08:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3847cc8c9b014efabee91e663e5700a3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/88977384b41b4c0b855dc96749c55bd2 : SUCCESS in 6m 30s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/fde582a12a724d8b866cfc023f610c87 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"78cf653652f861744775b9b39ad241822e360e9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 22:08:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"120a635bd83af1aa0c0059c3914882768a36959c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 22:24:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eec7afab98284e8ca6d29e16b893f082\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/1396413d6e6c443bbe911daf241e6622 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eceb8690f6ac7d0a87f544b6dce06120ff04f2b5","revisions":{"eceb8690f6ac7d0a87f544b6dce06120ff04f2b5":{"kind":"REWORK","_number":1,"created":"2024-05-29 21:38:41.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/79/920779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/920779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/920779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/920779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/920779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/920779/1"}}},"commit":{"parents":[{"commit":"3336b1f7aeae67994dac5d63dc372fee752271d4","subject":"Merge \"Retire devstack-gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3336b1f7aeae67994dac5d63dc372fee752271d4"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-29 21:35:55.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-29 21:35:55.000000000","tz":-420},"subject":"Chown the /opt/git repo cache to zuul:zuul","message":"Chown the /opt/git repo cache to zuul:zuul\n\nLatest git packages on Ubuntu (and possibly other locations in the\nfuture) don\u0027t allow locally cloning repos owned by a different user by\ndefault. Attempting to do so results in this error:\n\n  fatal: detected dubious ownership in repository at \u0027/opt/git/opendev.org/foo/bar/.git\u0027\n  To add an exception for this directory, call:\n\n      git config --global --add safe.directory /opt/git/opendev.org/foo/bar/.git\n  fatal: Could not read from remote repository.\n\n  Please make sure you have the correct access rights\n  and the repository exists.\n\nCurrently the /opt/git repos are owned by root:root. We expect that\nzuul will be the most common user to interact with these cached repos so\nwe chown to zuul:zuul in order to avoid these problems as much as\npossible. Any cases not using zuul will have to determine a path foward\nfor that special circumstances.\n\nChange-Id: I7cb21869bae42baed5027a9380f60762ab8944e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eceb8690f6ac7d0a87f544b6dce06120ff04f2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eceb8690f6ac7d0a87f544b6dce06120ff04f2b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
