)]}'
{"id":"openstack%2Fproject-config~920922","triplet_id":"openstack%2Fproject-config~master~Ic383f538c42307f2de552f17cfd13e1bd1e37924","project":"openstack/project-config","branch":"master","topic":"jeepyb-gerrit-39","hashtags":[],"change_id":"Ic383f538c42307f2de552f17cfd13e1bd1e37924","subject":"Update jeepyb to run Gerrit 3.9 image builds","status":"MERGED","created":"2024-05-31 16:16:20.000000000","updated":"2024-06-03 16:39:05.000000000","submitted":"2024-06-03 16:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920922-jeepyb-gerrit-39","meta_rev_id":"2cc651596e4f368963397c470d88c48c10228a61","_number":920922,"virtual_id_number":920922,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-03 16:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-06-03 16:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2024-06-03 16:30:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-31 16:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8b6f750d3b79dc60e29a896e3561550d7dee15d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-31 16:16:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83d81dc82d34342f96770ed6f3cfbe0b16590f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 16:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfece79a8b504ae0b8cb2388ade17796\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8953dcdca0e456098e0b15ab66f361a : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"651e5744020ab41aff4dfed94d0b3f4225977df7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-06-03 16:30:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae0afd152e80b4c870ff7d4b1684b0b698858773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 16:31:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a164918a361c1535668326bac41dff2e4ad85a6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 16:39:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae11f9a084a54df5aff5503438d01322\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f9be9955fcd647e59bb37880fb3b17c8 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2cc651596e4f368963397c470d88c48c10228a61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 16:39:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d74b70471cda079a99ad20477a8fc07b00719881","revisions":{"d74b70471cda079a99ad20477a8fc07b00719881":{"kind":"REWORK","_number":1,"created":"2024-05-31 16:16:20.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/22/920922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/920922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/920922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/920922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/920922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/920922/1"}}},"commit":{"parents":[{"commit":"7948bea52bcaf274ae20b1eafd10967c5eb9ea85","subject":"Merge \"Revert \"Temporary update IPE ACL\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7948bea52bcaf274ae20b1eafd10967c5eb9ea85"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-31 15:21:19.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-31 15:21:19.000000000","tz":-420},"subject":"Update jeepyb to run Gerrit 3.9 image builds","message":"Update jeepyb to run Gerrit 3.9 image builds\n\nThis is part of the post Gerrit 3.9 upgrade. We need to keep jeepyb in\nalignment with the state of our gerrit and have it trigger the correct\njobs.\n\nChange-Id: Ic383f538c42307f2de552f17cfd13e1bd1e37924\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d74b70471cda079a99ad20477a8fc07b00719881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d74b70471cda079a99ad20477a8fc07b00719881"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
