)]}'
{"id":"openstack%2Fproject-config~921725","triplet_id":"openstack%2Fproject-config~master~I46852ce6c71e61dbe19f7f00b9411d43232b8550","project":"openstack/project-config","branch":"master","topic":"drop-nebulous","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-07-13 15:05:15.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2024-07-13 15:05:15.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-07-12 12:07:09.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I46852ce6c71e61dbe19f7f00b9411d43232b8550","subject":"Mark NebulOuS as moved (retire on OpenDev)","status":"MERGED","created":"2024-06-11 06:26:26.000000000","updated":"2024-07-13 15:28:19.000000000","submitted":"2024-07-13 15:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"921725-drop-nebulous","meta_rev_id":"d03a4f37431d8fd74facc54f1ab045a23ed0fd3a","_number":921725,"virtual_id_number":921725,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-13 15:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-07-12 20:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-12 12:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-13 14:59:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-11 06:27:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-06-11 06:27:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-06-11 06:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"130e52087d4b49e57566dff5b6419ce1db2611dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-06-11 06:26:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba36ae4ca2c009624fe108031a703a25a20302d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 06:41:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ee23012f1eb41a39fb20187f53efb7f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/afa8ca2eb06b4ce2a9590de10ce6eeeb : SUCCESS in 6m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/db35e10b266b46f3aafb9a6249833c4a : FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e6d4b0b26d234b1fedeb11f882a747920407ef75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-06-11 07:04:45.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ad1d1792affabc3a31e1bc8288ace3796080e249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 07:18:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713a4e1e0110464aa461585f9c2f6eb5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f212aa42f8d3434aa3746fe8315fe1c4 : SUCCESS in 6m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8213f2e559b0439b89ea4e53004237cb : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"195a1898f5804bf2a836a0b1ed98bbadd7db6633","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-01 10:46:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9519a1afc188812e5e814dca74ae1143595eb667","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-11 17:37:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb458c4f46b12fbd4134d4bc750cb72c250d1bd3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-11 17:44:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4f58712b606b4bc198e2326b346ea5b015e62cb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-11 17:54:49.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"2e4eb0224f0337ccbd47f508e8cf8b0da3ddd5d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-07-11 18:00:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1afb9705902e5001cba605cbb73bd1278f0cb991","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-11 18:01:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b6c94c5a32704f884a5b44d06f84e2df9c5e2279","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-11 18:08:12.000000000","message":"Patch Set 3: Workflow-1\n\nSounds like this is going to be revised to switch affected repositories to the opendev/retired.config ACL.","accounts_in_message":[],"_revision_number":3},{"id":"9213567b86b10f64e06aaf883a8ce67cc7b2aaa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 18:13:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a00f4778b054dce89bf8eab10bca515\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d22a6c8e4e024fb2ae5a19c2ae958432 : SUCCESS in 6m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d68460a5a43b4121943d5523ca077dab : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"8bbaf8f6bf422d111f9e2dea4b36634afd8f1ce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-07-12 09:54:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ae78b7970788a2cd4d98b22b2da96dcd66a6fbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 10:15:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f1d1c32a96440c9110c9d256df1669\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/14ce9a28948f4d9bb9a7dd9e5a6ea684 : SUCCESS in 6m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ecfbe15c8ca4425fb6c728a602dc7580 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":4},{"id":"897b7c73b14200ff30e42f9ed594f9ee710b4360","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-07-12 10:18:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"362b0a436a8930c8e07772ad559bba79e3efa2c3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-12 12:07:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e6498e701078b44287d9198c6d7abd4d658fd862","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-12 20:31:28.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e828a3f5ca3e94459b91a4a8edaf8dbdbb2df7e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2024-07-13 12:53:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be18948a886445688e7fddc93ee806b9d31d1039","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-13 14:59:28.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"631478b887ed48b954062000a146a114e6a70c2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 14:59:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"785d52b3bc88ae7d85ad8582915915cc0ca75d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 15:05:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/670570cb8fc64e038a0a949c5a1fa226\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d1e595e23a14fa299c914cf1ebb597a : SUCCESS in 5m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2de1c46c042f40b18a3ebcc448a8514f : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"daedd7d37e81dfdf7b030ce0e4175d0814afe605","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 15:05:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d03a4f37431d8fd74facc54f1ab045a23ed0fd3a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 15:28:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3157e254141e4395941298fcc56687b7\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1e639890c9634d37b034e6b8d015ff1d : SUCCESS in 4m 22s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/927fa64b754d45e9959f40b077b468ff : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9a30108caef2a29c8327dcca774ee9c45f6ef3a2","revisions":{"efeb8ef13f084513338cbb8a352ed6a245c93b90":{"kind":"REWORK","_number":1,"created":"2024-06-11 06:26:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/921725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/921725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/921725/1"}}},"commit":{"parents":[{"commit":"d6ca6fd3f873941719eb6452bca84f9d59aa1071","subject":"Merge \"Add wandertracks repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6ca6fd3f873941719eb6452bca84f9d59aa1071"}]}],"author":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-06-11 06:24:40.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-06-11 06:24:40.000000000","tz":120},"subject":"Drop NebulOuS Zuul tenant","message":"Drop NebulOuS Zuul tenant\n\nDue to the migration of NebulOuS to GitHub and GitHub Actions,\nlet\u0027s switch NebulOuS Zuul tenant off.\n\nThis change also reconfigures Gerrit to allow core groups in\nNebulOuS to submit changes.\n\nChange-Id: I46852ce6c71e61dbe19f7f00b9411d43232b8550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efeb8ef13f084513338cbb8a352ed6a245c93b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efeb8ef13f084513338cbb8a352ed6a245c93b90"}]},"branch":"refs/heads/master"},"e1dff9ffac77d642938bd079105fbf7985ec844e":{"kind":"REWORK","_number":2,"created":"2024-06-11 07:04:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/921725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/921725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/921725/2"}}},"commit":{"parents":[{"commit":"d6ca6fd3f873941719eb6452bca84f9d59aa1071","subject":"Merge \"Add wandertracks repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6ca6fd3f873941719eb6452bca84f9d59aa1071"}]}],"author":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-06-11 06:24:40.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-06-11 07:04:40.000000000","tz":0},"subject":"Drop NebulOuS Zuul tenant","message":"Drop NebulOuS Zuul tenant\n\nDue to the migration of NebulOuS to GitHub and GitHub Actions,\nlet\u0027s switch NebulOuS Zuul tenant off.\n\nThis change also reconfigures Gerrit to allow core groups in\nNebulOuS to submit changes.\n\nChange-Id: I46852ce6c71e61dbe19f7f00b9411d43232b8550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1dff9ffac77d642938bd079105fbf7985ec844e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1dff9ffac77d642938bd079105fbf7985ec844e"}]},"branch":"refs/heads/master"},"f20de3cc3634a5522276882817490695ae44bc02":{"kind":"REWORK","_number":3,"created":"2024-07-11 18:00:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/921725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/921725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/921725/3"}}},"commit":{"parents":[{"commit":"d6ca6fd3f873941719eb6452bca84f9d59aa1071","subject":"Merge \"Add wandertracks repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6ca6fd3f873941719eb6452bca84f9d59aa1071"}]}],"author":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-06-11 06:24:40.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-07-11 17:59:44.000000000","tz":0},"subject":"Drop NebulOuS Zuul tenant","message":"Drop NebulOuS Zuul tenant\n\nDue to the migration of NebulOuS to GitHub and GitHub Actions,\nlet\u0027s switch NebulOuS Zuul tenant off.\n\nChange-Id: I46852ce6c71e61dbe19f7f00b9411d43232b8550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f20de3cc3634a5522276882817490695ae44bc02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f20de3cc3634a5522276882817490695ae44bc02"}]},"branch":"refs/heads/master"},"9a30108caef2a29c8327dcca774ee9c45f6ef3a2":{"kind":"REWORK","_number":4,"created":"2024-07-12 09:54:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/921725/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/921725/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/921725/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/921725/4"}}},"commit":{"parents":[{"commit":"dcf0ec603f592c08d6afc636bce8cf3ef3a7cda2","subject":"Ironic no longer uses storyboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dcf0ec603f592c08d6afc636bce8cf3ef3a7cda2"}]}],"author":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-06-11 06:24:40.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"rpiliszek@7bulls.com","date":"2024-07-12 09:54:48.000000000","tz":120},"subject":"Mark NebulOuS as moved (retire on OpenDev)","message":"Mark NebulOuS as moved (retire on OpenDev)\n\nFollowing [1] with adaptations.\n\nDue to the migration of NebulOuS to GitHub and GitHub Actions,\nlet\u0027s switch NebulOuS Zuul tenant off and mark repos properly.\n\n[1] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#step-3-remove-project-from-infrastructure-systems\n\nChange-Id: I46852ce6c71e61dbe19f7f00b9411d43232b8550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a30108caef2a29c8327dcca774ee9c45f6ef3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a30108caef2a29c8327dcca774ee9c45f6ef3a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
