)]}'
{"id":"openstack%2Fproject-config~921987","triplet_id":"openstack%2Fproject-config~master~If8e9b7105b4c7a13e87ebb4f6c985e821c30a842","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-06-13 22:15:20.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If8e9b7105b4c7a13e87ebb4f6c985e821c30a842","subject":"Add OpenMetal to Nodepool and Grafana","status":"MERGED","created":"2024-06-13 20:29:30.000000000","updated":"2024-06-13 22:42:40.000000000","submitted":"2024-06-13 22:33:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"921987","meta_rev_id":"04b641dfb703fcff05b49437826f3f4036ec4693","_number":921987,"virtual_id_number":921987,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-13 22:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-06-13 20:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-13 22:15:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-13 20:37:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-06-13 20:54:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74b41ad751fbcb855da684e1fd5adeec11d0a5ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-13 20:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab65d5c7bbe431c9bab6579674421792d2953ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-13 20:34:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ff8f24b9a29ac9713a87f27f01f6ba6c9cd8147","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-06-13 20:37:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41b24906f3025e700749f02ce7b053126a45673a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 20:54:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e0a950fc5384e5ba0e8c185c456f4de\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bd3d7d378aef49cc8ee00ed44307a2b6 : SUCCESS in 18m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42f9d295a2f44c93a3c70e7782a677ed : SUCCESS in 7m 31s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/77657ceea1b64c59aba22f365d88f6e8 : SUCCESS in 13m 23s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ee6e9fb1c1b44adeb04ab54bdd644567 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9d913661e37f842a7bbb287999f9433ddf45673f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-13 22:15:20.000000000","message":"Patch Set 2: Workflow+1\n\nSelf-approving to keep this moving, will watch and make sure images upload after it deploys.","accounts_in_message":[],"_revision_number":2},{"id":"dfa537c97b9009baf8ef04c30773eb900ffd172b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 22:15:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b381880d71edda54e29819f631a337eb525817b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 22:33:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ffbb7c3532e496c9de6a8a98d8bd6b2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e7e3aeb728bd4fdbb008eae3aeff8d34 : SUCCESS in 15m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/388acdae197042449fa1ae47582739df : SUCCESS in 5m 03s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/be29c6153285474896a9d9562bd338aa : SUCCESS in 11m 37s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2604b15ae69a4008b04a896cf102a7e1 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"525cbe33cd62ceafc59776e65d0a1cfeda558c74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 22:33:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"04b641dfb703fcff05b49437826f3f4036ec4693","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 22:42:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8edb4ab8fc3b47d49ca622d4a3fd25af\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/dc048124ab1546abac9751916f0e42ba : SUCCESS in 6m 41s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/bf1c9ac8e59741a9a4928a4e35b18419 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28ae63164225a914a89cc7a057f422d7b9ca1685","revisions":{"cfd03759ac9c291d17da5663ccecf23edec109f8":{"kind":"REWORK","_number":1,"created":"2024-06-13 20:29:30.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/87/921987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/87/921987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/921987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/921987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/921987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/87/921987/1"}}},"commit":{"parents":[{"commit":"cea490f471c339cc5d454b139e65dadd3613f77a","subject":"Pause centos-8-stream image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cea490f471c339cc5d454b139e65dadd3613f77a"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-06-13 20:13:23.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-06-13 20:29:08.000000000","tz":0},"subject":"Add OnMetal to Nodepool and Grafana","message":"Add OnMetal to Nodepool and Grafana\n\nThis is essentially reverting commits bd15ddc and cb4b99b which were\nthe final stages of winding down and cleaning up the old InMotion\ncloud which OnMetal has replaced, with the cloud name updated (but\nregion kept the same) and grafyaml data regenerated. It stops short\nof actually booting nodes in the new environment until we have a\nchance to spot check things once images get uploaded.\n\nSince this is re-adding diskimages back to nl02, I refrained from\nincluding centos-8-stream which is in the progress of being removed,\nso that we don\u0027t unnecessarily upload images we\u0027re not planning to\nboot.\n\nChange-Id: If8e9b7105b4c7a13e87ebb4f6c985e821c30a842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cfd03759ac9c291d17da5663ccecf23edec109f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cfd03759ac9c291d17da5663ccecf23edec109f8"}]},"branch":"refs/heads/master"},"28ae63164225a914a89cc7a057f422d7b9ca1685":{"kind":"REWORK","_number":2,"created":"2024-06-13 20:34:21.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/87/921987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/87/921987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/921987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/921987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/87/921987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/87/921987/2"}}},"commit":{"parents":[{"commit":"cea490f471c339cc5d454b139e65dadd3613f77a","subject":"Pause centos-8-stream image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cea490f471c339cc5d454b139e65dadd3613f77a"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-06-13 20:13:23.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-06-13 20:34:06.000000000","tz":0},"subject":"Add OpenMetal to Nodepool and Grafana","message":"Add OpenMetal to Nodepool and Grafana\n\nThis is essentially reverting commits bd15ddc and cb4b99b which were\nthe final stages of winding down and cleaning up the old InMotion\ncloud which OpenMetal has replaced, with the cloud name updated (but\nregion kept the same) and grafyaml data regenerated. It stops short\nof actually booting nodes in the new environment until we have a\nchance to spot check things once images get uploaded.\n\nSince this is re-adding diskimages back to nl02, I refrained from\nincluding centos-8-stream which is in the progress of being removed,\nso that we don\u0027t unnecessarily upload images we\u0027re not planning to\nboot.\n\nChange-Id: If8e9b7105b4c7a13e87ebb4f6c985e821c30a842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28ae63164225a914a89cc7a057f422d7b9ca1685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/28ae63164225a914a89cc7a057f422d7b9ca1685"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
