)]}'
{"id":"openstack%2Fproject-config~922357","triplet_id":"openstack%2Fproject-config~master~If5581b4531d0395bd4b484047bdc81666e2c2bfd","project":"openstack/project-config","branch":"master","topic":"drop-centos-8s","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-06-20 01:08:40.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"If5581b4531d0395bd4b484047bdc81666e2c2bfd","subject":"Remove publish-wheel-cache-centos-8-stream* jobs","status":"ABANDONED","created":"2024-06-20 00:27:07.000000000","updated":"2024-06-20 01:43:28.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19e602884bc65d66510aea5bf0b1aafc96790f20","_number":922357,"virtual_id_number":922357,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-20 01:15:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-20 00:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"960d0cba53d13aa0723db2514f33cae03c390468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-20 00:27:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0265960f8cc7f789cc8837b5311b88e7a127c76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 00:27:37.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  Job publish-wheel-cache-centos-8-stream not defined\n\nThe problem appears in the the \"opendev.org/openstack/requirements\" project stanza:\n\n  project:\n      name: openstack/requirements\n      queue: integrated\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n      check:\n        jobs:\n          - openstack-tox-docs:\n  ...\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4082","accounts_in_message":[],"_revision_number":1},{"id":"0dd1ea16eaa3e77b779232cc48b6dbdfbf156ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-20 01:03:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cf7d80f395d0a764ca1957dbd575727d91c3cd85","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-20 01:08:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09407631b0ec7f3aaf40dbc3bce4b17ecd7555e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 01:15:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e905e77e92bf44b7b15d4c5d2328a3e9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b84e89f4022749c0bea7e3edb75d5568 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"19e602884bc65d66510aea5bf0b1aafc96790f20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-20 01:43:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6698a65f85292d1a2a98918b702c3132f4708496","revisions":{"0f18abf25e11f7a10ee9c883cf52149534c89597":{"kind":"REWORK","_number":1,"created":"2024-06-20 00:27:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/57/922357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/922357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/922357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/922357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/922357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/922357/1"}}},"commit":{"parents":[{"commit":"f36b8cf6a9d52ef9c5256918e9839cb3d4f92391","subject":"Use the \"nova\" AZ for Nodepool in openmetal-iad3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f36b8cf6a9d52ef9c5256918e9839cb3d4f92391"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-06-20 00:24:58.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-06-20 00:25:55.000000000","tz":600},"subject":"Remove publish-wheel-cache-centos-8-stream* jobs","message":"Remove publish-wheel-cache-centos-8-stream* jobs\n\nPlatform is largely gone we can\u0027t really build wheels, so remove it\n\nChange-Id: If5581b4531d0395bd4b484047bdc81666e2c2bfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f18abf25e11f7a10ee9c883cf52149534c89597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f18abf25e11f7a10ee9c883cf52149534c89597"}]},"branch":"refs/heads/master"},"6698a65f85292d1a2a98918b702c3132f4708496":{"kind":"REWORK","_number":2,"created":"2024-06-20 01:03:36.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/57/922357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/922357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/922357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/922357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/922357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/922357/2"}}},"commit":{"parents":[{"commit":"f36b8cf6a9d52ef9c5256918e9839cb3d4f92391","subject":"Use the \"nova\" AZ for Nodepool in openmetal-iad3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f36b8cf6a9d52ef9c5256918e9839cb3d4f92391"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-06-20 00:24:58.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-06-20 00:59:19.000000000","tz":600},"subject":"Remove publish-wheel-cache-centos-8-stream* jobs","message":"Remove publish-wheel-cache-centos-8-stream* jobs\n\nPlatform is largely gone we can\u0027t really build wheels, so remove it\n\nChange-Id: If5581b4531d0395bd4b484047bdc81666e2c2bfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6698a65f85292d1a2a98918b702c3132f4708496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6698a65f85292d1a2a98918b702c3132f4708496"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
