)]}'
{"id":"openstack%2Fproject-config~922864","triplet_id":"openstack%2Fproject-config~master~Ic843113c48c109ff6e55acc4252da41176fcc5e6","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-06-27 21:57:56.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-06-27 21:50:47.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic843113c48c109ff6e55acc4252da41176fcc5e6","subject":"Ironic no longer uses storyboard","status":"MERGED","created":"2024-06-26 18:38:40.000000000","updated":"2024-06-27 22:02:08.000000000","submitted":"2024-06-27 21:57:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"922864","meta_rev_id":"be306b92d6c428c21b12b2f8238f0e7117fb3f3e","_number":922864,"virtual_id_number":922864,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 21:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-06-27 20:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-27 21:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-27 21:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 18:46:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-26 18:50:47.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2024-06-26 19:25:16.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-06-26 20:05:24.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2024-06-27 20:41:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"4c23eb1acbc6639da9ce1a9eedbd8bf2c57a8aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-06-26 18:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5db929a5b45323d2ac3b22cc203a6b3a9b4122a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:46:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2897022b48dc49548d5bde74abab10d6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/25e72f870c86478c89e9862ed6b4d699 : SUCCESS in 6m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/923e6ff60b7848058d97908ac4b8451f : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"136797e6612a7f21554efd614c618941dc6680d4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-26 18:50:47.000000000","message":"Patch Set 1:\n\nTo be clear, when not enabling use-storyboard, the groups directive specifies the project\u0027s name on Launchpad (rather than defaulting to the same name as the repository). So in this case, it indicates all affected projects will use https://bugs.launchpad.net/ironic for defect tracking. If that\u0027s the intent, then this change looks right to me.","accounts_in_message":[],"_revision_number":1},{"id":"0768fae37b6ae994272fed06b14c2acd6014ab15","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-26 18:52:11.000000000","message":"Patch Set 1:\n\nOh, except for the actual openstack/ironic project entry, which should not specify the ironic group as it would be redundant for Launchpad.","accounts_in_message":[],"_revision_number":1},{"id":"511b893ead5b83b0d931de09b96d8ea14afd4ea7","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-06-26 18:52:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de95b6b4310600215b58e46ab5be29cd28e8a50a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-26 19:01:35.000000000","message":"Patch Set 1:\n\nCorrect, sorry, the groups option is for maintaining project groups in StoryBoard or for overriding the default project name in Launchpad. There\u0027s not really any additional metadata or commentary in the file, though perhaps if \"part of Ironic\" is important enough to convey it could be worked into all of the project descriptions?","accounts_in_message":[],"_revision_number":1},{"id":"566b7f15d548d7c6326733eafb86b8c6059e6cc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-06-26 19:10:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce1a7dd147fb86ef872906ca8963cb42f6dbf22d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-06-26 19:10:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41efbe3f1b946201469629c36b3d41f76e913f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 19:20:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518102cabac24c53b5b44a14085d861c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d2e9d6b45b164de79ab2babeeb5da761 : SUCCESS in 7m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/551e1c16d0964c629ddffd0d1c994e36 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"09c5e6e4b4e252cdc43bc0182e1f0a34c94c6105","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-26 19:25:16.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, but note that I have not manually checked whether https://bugs.launchpad.net/$PROJECT exists for each of these.","accounts_in_message":[],"_revision_number":2},{"id":"7c2cead54976fb93ad19fd37cc07b10fb69e6c35","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-06-26 20:05:24.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6beb181c0ef87fe37e7a5c5124cc32df2069f319","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-06-26 20:07:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad23d535d5fa4ee8e78ed98b0fdbd05183facc00","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-06-26 22:48:27.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79a41b83a486a5ba6ae37e8f8fdeed2460da1647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-06-27 20:31:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7e0580172fb7f4cf98fea7b26b8eb1a1f64d0393","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-06-27 20:41:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"abdacc968e39061f3ee6e560f906b24e8160b9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:44:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68bc089b6b954245a5e0316fef3036da\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/97de13c279174162abd5a6d75a22401a : SUCCESS in 7m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/83e074f5d4b54fd5a86e412a3d7afd64 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"31c461d9664a60612a56fb3673f3f095fdfdf4c2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-06-27 21:50:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a8c9742b8a3e0931c88a67d90def81124020aa86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:51:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47c588cde5facac2ef9004ebc83b0e74ae619596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:57:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca3f429c80044abb25dfa50bf18cfac\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2213304afdbb49e091e42902c3005a8c : SUCCESS in 6m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b806210d7769489bad9330b18a45fc55 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e3e933b41e032e6772adf17d47033dceaecbc4c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 21:57:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"be306b92d6c428c21b12b2f8238f0e7117fb3f3e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 22:02:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c483e44ffc241b39ac15ed3cd55d1ca\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/7d03e6dcb09d411a843fa3ec75f67601 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dcf0ec603f592c08d6afc636bce8cf3ef3a7cda2","revisions":{"0172b141a70bb96af0cf9cbb1f207422a0121628":{"kind":"REWORK","_number":1,"created":"2024-06-26 18:38:40.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/64/922864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/922864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/922864/1"}}},"commit":{"parents":[{"commit":"427e4b054eb279527f9a13e496a972900eb49c47","subject":"Merge \"Drop wheel publishing for centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/427e4b054eb279527f9a13e496a972900eb49c47"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-06-26 18:37:44.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-06-26 18:37:44.000000000","tz":-420},"subject":"Ironic no longer uses storyboard","message":"Ironic no longer uses storyboard\n\n- Update all ironic-related projects to no longer advertise storyboard\n  use\n- Add proper tag to metalsmith indicating it\u0027s a project in the ironic\n  group\n\nCloses-bug: #2056146\nChange-Id: Ic843113c48c109ff6e55acc4252da41176fcc5e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0172b141a70bb96af0cf9cbb1f207422a0121628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0172b141a70bb96af0cf9cbb1f207422a0121628"}]},"branch":"refs/heads/master"},"c1238d0992c408a8cf2b43848e03b3ad09fb896e":{"kind":"REWORK","_number":2,"created":"2024-06-26 19:10:14.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/64/922864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/922864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/922864/2"}}},"commit":{"parents":[{"commit":"427e4b054eb279527f9a13e496a972900eb49c47","subject":"Merge \"Drop wheel publishing for centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/427e4b054eb279527f9a13e496a972900eb49c47"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-06-26 18:37:44.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-06-26 19:10:09.000000000","tz":-420},"subject":"Ironic no longer uses storyboard","message":"Ironic no longer uses storyboard\n\n- Update all ironic-related projects to no longer advertise storyboard\n  use\n- Add proper tag to metalsmith indicating it\u0027s a project in the ironic\n  group\n\nCloses-bug: #2056146\nChange-Id: Ic843113c48c109ff6e55acc4252da41176fcc5e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1238d0992c408a8cf2b43848e03b3ad09fb896e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1238d0992c408a8cf2b43848e03b3ad09fb896e"}]},"branch":"refs/heads/master"},"dcf0ec603f592c08d6afc636bce8cf3ef3a7cda2":{"kind":"REWORK","_number":3,"created":"2024-06-27 20:31:38.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/64/922864/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/922864/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/922864/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/922864/3"}}},"commit":{"parents":[{"commit":"427e4b054eb279527f9a13e496a972900eb49c47","subject":"Merge \"Drop wheel publishing for centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/427e4b054eb279527f9a13e496a972900eb49c47"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-06-26 18:37:44.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-06-27 20:31:32.000000000","tz":-420},"subject":"Ironic no longer uses storyboard","message":"Ironic no longer uses storyboard\n\n- Update all ironic-related projects to no longer advertise storyboard\n  use\n- Add proper tag to metalsmith indicating it\u0027s a project in the ironic\n  group\n\nCloses-bug: #2056146\nChange-Id: Ic843113c48c109ff6e55acc4252da41176fcc5e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dcf0ec603f592c08d6afc636bce8cf3ef3a7cda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dcf0ec603f592c08d6afc636bce8cf3ef3a7cda2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
