)]}'
{"id":"openstack%2Fproject-config~924430","triplet_id":"openstack%2Fproject-config~master~I23b445164f8ad82310c0536c29679368631b7616","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-07-26 22:52:13.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-07-26 19:40:07.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-07-26 22:52:13.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-07-26 20:08:14.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I23b445164f8ad82310c0536c29679368631b7616","subject":"Implement manila-unmaintained-core group","status":"MERGED","created":"2024-07-18 18:21:45.000000000","updated":"2024-07-26 22:56:42.000000000","submitted":"2024-07-26 22:52:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"924430","meta_rev_id":"b36c66f503ae7053c076b0faea4476ca1b64322c","_number":924430,"virtual_id_number":924430,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-26 22:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-07-26 22:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2024-07-26 18:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-26 20:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2024-07-26 22:39:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-18 18:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-18 18:36:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-07-26 18:07:26.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-07-26 19:40:07.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-07-26 19:40:07.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-07-26 22:39:15.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"bd571d3a2af4290337b50e50c00b96bac7d9198f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-18 18:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe2abd284dcea5e509fd5a3db642e941d579c0dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 18:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d52e9bf333fa481c97c9696d9629b58e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f6f07b3d120f4a349f1d5a568759d781 : SUCCESS in 6m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dff45d4cb82744d7ac2e4a89031bda8e : FAILURE in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"56201aa96e32871c4eb7cb4469244e3f51acc3a9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-18 18:36:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c2f0756952c503af5299a3e4416d8c715296826","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-18 18:37:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df7e78e42ca9d476e7df4ebfaeef464dce8efb5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-18 18:42:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"869a6078c39dd5625392d48715b91480579048e2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-18 18:43:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9dee8dbf1a033a4517fd5df59b7f5ec193d223d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-18 18:45:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98374db479ec25bd125d81fa0020ced415860c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 18:53:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e3db787954d479db747e5f641a254e2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c748688e5d54234aef75d7ef29ed232 : SUCCESS in 6m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a6eaa82977574ba8bca7d4505f659f4c : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"0af2ff42579e4a67ef53448e9670a558ec84ae1a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-18 19:04:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b8a206b9e7848f63c7b95589f0df52d19758534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-26 18:05:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f761d5666f51688369afc1efdff198b584891aa5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-26 18:06:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b9ca067daa709ed4bc89900c8e9df223a5f44aef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-26 18:07:26.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4622d6e1771bc24e6b540ad92cea4a29220c2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 18:14:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8664a717eca64358a71387a9eb310b30\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7bf01a21decd4e279915c2b16fcd0ab8 : SUCCESS in 6m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/19c8c29cc1ba44e1833224fe7ae51075 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"397d0e1a8a5a492698a7c6cc6192ec5d51c01fb4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-26 20:08:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"05aea1dcfa90eff9c2b1a93f20ff465e965a294c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-26 22:39:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"faa8a4fe42275051095594d6759ef19df6a3c7ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 22:39:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b5e900664c26355f9fa2f8592017eddbc0e560b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 22:52:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9150fa360444e8cb710af6431d08da8\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8d0842018bad4c18bc12a6cd31149d71 : SUCCESS in 6m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/12516fb6a2ff4483974505811c0057fe : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"ec5147c476d29d3a0046c42e5011c2683452211c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 22:52:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b36c66f503ae7053c076b0faea4476ca1b64322c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 22:56:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf00be8204c443bb333d5272eff2732\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/4450dfd19fef4d5c99c19b72380b0465 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bfeb741a66cd8f3d6a397141e9608f6317198c16","revisions":{"42f62c11e1cf0e99888feaa4d9d724984a5ee587":{"kind":"REWORK","_number":1,"created":"2024-07-18 18:21:45.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/30/924430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/924430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/924430/1"}}},"commit":{"parents":[{"commit":"2086ea3036b898d316b10870062c9b46db00ad16","subject":"Merge \"Mark NebulOuS as moved (retire on OpenDev)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2086ea3036b898d316b10870062c9b46db00ad16"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:20:39.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:21:42.000000000","tz":-180},"subject":"Implement manila-unmaintained-core group","message":"Implement manila-unmaintained-core group\n\nDefine permissions for manila-unmaintained-core, which will\noverride the global openstack-unmaintained-core for manila\nand manila-ui repositories, as allowed by TC resolution 2023-11-14.\n\nChange-Id: I23b445164f8ad82310c0536c29679368631b7616\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/42f62c11e1cf0e99888feaa4d9d724984a5ee587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/42f62c11e1cf0e99888feaa4d9d724984a5ee587"}]},"branch":"refs/heads/master"},"25c14e20a3abcdaec065e258ffb2b87f3306da9c":{"kind":"REWORK","_number":2,"created":"2024-07-18 18:42:57.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/30/924430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/924430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/924430/2"}}},"commit":{"parents":[{"commit":"2086ea3036b898d316b10870062c9b46db00ad16","subject":"Merge \"Mark NebulOuS as moved (retire on OpenDev)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2086ea3036b898d316b10870062c9b46db00ad16"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:20:39.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:42:49.000000000","tz":-180},"subject":"Implement manila-unmaintained-core group","message":"Implement manila-unmaintained-core group\n\nDefine permissions for manila-unmaintained-core, which will\noverride the global openstack-unmaintained-core for manila\nand manila-ui repositories, as allowed by TC resolution 2023-11-14.\n\nChange-Id: I23b445164f8ad82310c0536c29679368631b7616\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25c14e20a3abcdaec065e258ffb2b87f3306da9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25c14e20a3abcdaec065e258ffb2b87f3306da9c"}]},"branch":"refs/heads/master"},"a65f218fd7ea40dab1cc7ac31fb2a62793311557":{"kind":"REWORK","_number":3,"created":"2024-07-18 18:45:16.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/30/924430/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/924430/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/924430/3"}}},"commit":{"parents":[{"commit":"2086ea3036b898d316b10870062c9b46db00ad16","subject":"Merge \"Mark NebulOuS as moved (retire on OpenDev)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2086ea3036b898d316b10870062c9b46db00ad16"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:20:39.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:45:08.000000000","tz":-180},"subject":"Implement manila-unmaintained-core group","message":"Implement manila-unmaintained-core group\n\nDefine permissions for manila-unmaintained-core, which will\noverride the global openstack-unmaintained-core for manila\nand manila-ui repositories, as allowed by TC resolution 2023-11-14.\n\nChange-Id: I23b445164f8ad82310c0536c29679368631b7616\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a65f218fd7ea40dab1cc7ac31fb2a62793311557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a65f218fd7ea40dab1cc7ac31fb2a62793311557"}]},"branch":"refs/heads/master"},"bfeb741a66cd8f3d6a397141e9608f6317198c16":{"kind":"REWORK","_number":4,"created":"2024-07-26 18:05:24.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/30/924430/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/924430/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/924430/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/924430/4"}}},"commit":{"parents":[{"commit":"2086ea3036b898d316b10870062c9b46db00ad16","subject":"Merge \"Mark NebulOuS as moved (retire on OpenDev)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2086ea3036b898d316b10870062c9b46db00ad16"}]}],"author":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-18 18:20:39.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-07-26 18:05:16.000000000","tz":-180},"subject":"Implement manila-unmaintained-core group","message":"Implement manila-unmaintained-core group\n\nDefine permissions for manila-unmaintained-core, which will\noverride the global openstack-unmaintained-core for manila\nand manila-ui repositories, as allowed by TC resolution 2023-11-14.\n\nChange-Id: I23b445164f8ad82310c0536c29679368631b7616\nSigned-off-by: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfeb741a66cd8f3d6a397141e9608f6317198c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfeb741a66cd8f3d6a397141e9608f6317198c16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
