)]}'
{"id":"openstack%2Fproject-config~924791","triplet_id":"openstack%2Fproject-config~master~I71421ce9a10438549ef21441349be84b5d7bd38b","project":"openstack/project-config","branch":"master","topic":"drop-centos-8-stream","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-07-30 22:46:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71421ce9a10438549ef21441349be84b5d7bd38b","subject":"Stop building centos 8 stream images","status":"MERGED","created":"2024-07-23 20:57:37.000000000","updated":"2024-07-30 22:53:24.000000000","submitted":"2024-07-30 22:46:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924791-drop-centos-8-stream","meta_rev_id":"12b215c0dc91938932662fd2152534f76d674da8","_number":924791,"virtual_id_number":924791,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 22:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-30 21:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-30 22:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 20:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-30 21:21:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"908af3b20541985f003b38a257193453c0e1246a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-23 20:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94d8da232d72a01c851d460a697107d793f7af66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 20:58:11.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"b840845171913da12c4a61c64f8764d21ef47f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-30 21:02:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6d9da56e73c73d62cb28281d954db3d1a834c22e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 21:08:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0047a0e8e58742edafd163217c62f8be\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64e957d712494fb39fd44013ef3a6494 : SUCCESS in 6m 15s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/7cce78145a1840c1af1a0c9e60fd47f7 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"cefa0e380044880fa233e6eb79e64344da3db29b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-30 21:21:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"685ed12a962ba1e2aeb3bf1094ee2bdb2bfd21c6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-30 22:39:51.000000000","message":"Patch Set 2: Workflow+1\n\nUploaded images have been automatically deleted from all our providers due to the parent change, so this should now be safe to merge.","accounts_in_message":[],"_revision_number":2},{"id":"542e4c4956dcde6b34ec621c0e164bc0e713cf20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 22:40:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84b2e231803baa96a96aa7b49b3272653bfede84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 22:46:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c1425ffa0d42c38a51e697b7d7ea7e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/89a2d348c9834b609b7a0ab9d2be6b8f : SUCCESS in 6m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/85e5806fd4e8481ab77b8ae24847d33c : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"abdd39d3ebbad67b8f2fa148e81d182c50a948ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 22:46:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"12b215c0dc91938932662fd2152534f76d674da8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 22:53:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92f0151e054f42948e0a79737c4bdaf8\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/f405610268374825bd000204ab34ec84 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ebb5b0cb7e219874bbff0c2c5b3893f539c735b1","revisions":{"486e67ab09557a96ed57660f49d687dca9808407":{"kind":"REWORK","_number":1,"created":"2024-07-23 20:57:37.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/91/924791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/924791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/924791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/924791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/924791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/924791/1"}}},"commit":{"parents":[{"commit":"01cc594fc6e9dd911a3f32569f147793c39889b1","subject":"Remove centos-8-stream image uploads and labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01cc594fc6e9dd911a3f32569f147793c39889b1"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-23 20:56:41.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-23 20:56:41.000000000","tz":-420},"subject":"Stop building centos 8 stream images","message":"Stop building centos 8 stream images\n\nThis should be the last piece of cleanup to remove CentOS 8 stream from\nNodepool. We basically tell the builders to forget about these images\nwhich should result in cleanups of all the related records on disk and\nin zk.\n\nChange-Id: I71421ce9a10438549ef21441349be84b5d7bd38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/486e67ab09557a96ed57660f49d687dca9808407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/486e67ab09557a96ed57660f49d687dca9808407"}]},"branch":"refs/heads/master"},"ebb5b0cb7e219874bbff0c2c5b3893f539c735b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-30 21:02:02.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/91/924791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/924791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/924791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/924791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/924791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/924791/2"}}},"commit":{"parents":[{"commit":"66fbfb941e34fa723f1725ef6c813d9412c94e13","subject":"Remove centos-8-stream image uploads and labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66fbfb941e34fa723f1725ef6c813d9412c94e13"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-23 20:56:41.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-30 21:01:53.000000000","tz":-420},"subject":"Stop building centos 8 stream images","message":"Stop building centos 8 stream images\n\nThis should be the last piece of cleanup to remove CentOS 8 stream from\nNodepool. We basically tell the builders to forget about these images\nwhich should result in cleanups of all the related records on disk and\nin zk.\n\nChange-Id: I71421ce9a10438549ef21441349be84b5d7bd38b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebb5b0cb7e219874bbff0c2c5b3893f539c735b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebb5b0cb7e219874bbff0c2c5b3893f539c735b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
