)]}'
{"id":"openstack%2Fproject-config~925050","triplet_id":"openstack%2Fproject-config~master~I40acd28146484cb81f87ee825787d578eb964de8","project":"openstack/project-config","branch":"master","topic":"shutdown-linaro-cloud","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-07-30 16:16:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40acd28146484cb81f87ee825787d578eb964de8","subject":"Remove linaro cloud from Nodepool","status":"MERGED","created":"2024-07-26 22:00:12.000000000","updated":"2024-07-30 16:27:29.000000000","submitted":"2024-07-30 16:16:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925050-shutdown-linaro-cloud","meta_rev_id":"ee6a220555e65f8ce76bc6b5b3ca9cd487164e77","_number":925050,"virtual_id_number":925050,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 16:16:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2024-07-29 16:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-29 19:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-30 15:55:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-26 22:22:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-29 16:04:22.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2024-07-29 19:54:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"f743f96cba5a2af628685afda56d2f255dc7b7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-26 22:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"149f9c8d3ab8fc592f7bae2d2788d10e9deeef5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 22:22:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851c16127e0c431c91cb03bf25e33405\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/889255cb265f417489fc3f0ddefefb6c : SUCCESS in 13m 37s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/38192284c35b4f42bd76c10e70404f97 : SUCCESS in 6m 20s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/b577b6d9b07e4c6e93671b4ec4e7b6f1 : SUCCESS in 10m 47s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d6db3a3d8b774fed8cf81cd635de45a3 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8556589e4be235fd28f78542c01f4945e552560b","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-07-29 16:04:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"802e6b3cfbda6f9e68675fba97ff819039eba2c2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-29 19:54:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c657d886b45fd6318a82ea02d2e6f3ff158515d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-30 15:55:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6f3adc1cbfc09db4cec518c92df903f53514402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:55:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75fe9e074ccbfc8c253c9ab409d02959b9c75e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:16:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4b1857bc5014172ba58c4645cb286bb\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b0cfb33620d64c17b959e2d0e225e5c7 : SUCCESS in 21m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f40b1329d8e44a89d5710f610d9b11a : SUCCESS in 4m 58s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/46bbea83096043b99b18f464fc18de89 : SUCCESS in 12m 35s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3d8c16b7f25543be854c6e987e3d9431 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"60e077b5908d507b9f51c48143f46be93dd1cac5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:16:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee6a220555e65f8ce76bc6b5b3ca9cd487164e77","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:27:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0b1a8f1e8a4a15add32e17021e1c1a\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/42f0a9b8c53a4ee8a25ce070dd9e03a9 : SUCCESS in 5m 28s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/df0de0a093b84919b7b527238dc8893c : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbd79385961c6b9571d488de641e18205d9b5163","revisions":{"cbd79385961c6b9571d488de641e18205d9b5163":{"kind":"REWORK","_number":1,"created":"2024-07-26 22:00:12.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/50/925050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/925050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/925050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/925050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/925050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/925050/1"}}},"commit":{"parents":[{"commit":"3801f3530c2ba04e413a66aea760d2f06fd4ac7b","subject":"Remove labels and diskimages from the linaro cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3801f3530c2ba04e413a66aea760d2f06fd4ac7b"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-26 21:58:53.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-26 21:58:53.000000000","tz":-420},"subject":"Remove linaro cloud from Nodepool","message":"Remove linaro cloud from Nodepool\n\nThis should be done after we are happy with the cleanups that nodepool\nhas performed particularly against the zk database. Otherwise we may\nneed to do manual cleanup after this provider config is removed.\n\nWe also remove the auto generation of the linaro cloud grafana dashboard\nrules but don\u0027t remove the dashboard itself. We typically keep the\ndashboards around for a while as the information exposed there may\ncontinue to be useful after the cloud is gone.\n\nChange-Id: I40acd28146484cb81f87ee825787d578eb964de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cbd79385961c6b9571d488de641e18205d9b5163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cbd79385961c6b9571d488de641e18205d9b5163"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
