)]}'
{"id":"openstack%2Fproject-config~926949","triplet_id":"openstack%2Fproject-config~master~I602beef5cbed32d411536f89e89bffd0752bc7ca","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"37082":{"account":{"_account_id":37082,"name":"Maksym Medvied","display_name":"Maksym Medvied","email":"maksym.medvied@canonical.com","username":"medvied"},"last_update":"2024-08-22 20:41:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I602beef5cbed32d411536f89e89bffd0752bc7ca","subject":"Mirror charm-ceph-nfs to GitHub","status":"MERGED","created":"2024-08-22 19:18:30.000000000","updated":"2024-08-22 20:41:11.000000000","submitted":"2024-08-22 20:41:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926949","meta_rev_id":"0984319f23fb7dca1de8b591c14a27027726172f","_number":926949,"virtual_id_number":926949,"owner":{"_account_id":37082,"name":"Maksym Medvied","display_name":"Maksym Medvied","email":"maksym.medvied@canonical.com","username":"medvied"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 20:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-22 20:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-22 20:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 19:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 20:23:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"cdf8d2b8e27257fae340c265cda18d0ab570a42d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37082,"name":"Maksym Medvied","display_name":"Maksym Medvied","email":"maksym.medvied@canonical.com","username":"medvied"},"date":"2024-08-22 19:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2790db10e5cfb2603a5f18bb6776e124c3ad5a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 19:31:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fe3249bd54441886c96701eea5c0e1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e7c43b7395c041819b98101c63cb31fb : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1c83c7483192c4acd917ce076f6bd501b8c55322","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-08-22 20:23:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dd04b6c5be85c42f44e2d7d0c44929462b867a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 20:24:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6d386706aac85c9eb4e9e30d4d9dfaa8eecfe98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 20:41:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fc5e5b5ab304a04884a6e4d64bbc5a0\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9567c7d175ea4cc78f1a4e4ae655d470 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0984319f23fb7dca1de8b591c14a27027726172f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 20:41:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fd9b7f689cdcffcb50bc9ff3cfa07685ee82ae7","revisions":{"6fd9b7f689cdcffcb50bc9ff3cfa07685ee82ae7":{"kind":"REWORK","_number":1,"created":"2024-08-22 19:18:30.000000000","uploader":{"_account_id":37082,"name":"Maksym Medvied","display_name":"Maksym Medvied","email":"maksym.medvied@canonical.com","username":"medvied"},"ref":"refs/changes/49/926949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/49/926949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/926949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/926949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/49/926949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/49/926949/1"}}},"commit":{"parents":[{"commit":"bb686e8deb56ab17ea94f1ddbc4a07f4c8b89db5","subject":"[release-tool] Workaround for getting dist_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb686e8deb56ab17ea94f1ddbc4a07f4c8b89db5"}]}],"author":{"name":"Maksym Medvied","email":"maksym.medvied@canonical.com","date":"2024-08-22 19:17:58.000000000","tz":-240},"committer":{"name":"Maksym Medvied","email":"maksym.medvied@canonical.com","date":"2024-08-22 19:17:58.000000000","tz":-240},"subject":"Mirror charm-ceph-nfs to GitHub","message":"Mirror charm-ceph-nfs to GitHub\n\nChange-Id: I602beef5cbed32d411536f89e89bffd0752bc7ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6fd9b7f689cdcffcb50bc9ff3cfa07685ee82ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6fd9b7f689cdcffcb50bc9ff3cfa07685ee82ae7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
