)]}'
{"id":"openstack%2Fproject-config~927433","triplet_id":"openstack%2Fproject-config~master~Id88f3906a157a04b097b162950dedccc12a8fa90","project":"openstack/project-config","branch":"master","topic":"re-add-networking-cisco","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-28 22:22:56.000000000","reason":"Change was submitted"},"32213":{"account":{"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},"last_update":"2024-08-28 21:45:04.000000000","reason":"\u003cGERRIT_ACCOUNT_32213\u003e replied on the change","reason_account":{"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"}},"18031":{"account":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"last_update":"2024-08-28 22:22:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id88f3906a157a04b097b162950dedccc12a8fa90","subject":"Re-introduce gate jobs for networking-cisco","status":"MERGED","created":"2024-08-28 20:43:14.000000000","updated":"2024-08-28 22:29:49.000000000","submitted":"2024-08-28 22:22:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927433-re-add-networking-cisco","meta_rev_id":"e9f8286fd60c12c1ca5df1672502e3760f57accc","_number":927433,"virtual_id_number":927433,"owner":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 22:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},"all":[{"value":1,"date":"2024-08-28 21:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 22:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 22:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-28 20:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-28 20:44:03.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-08-28 21:38:48.000000000","updated_by":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"reviewer":{"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},"state":"REVIEWER"},{"updated":"2024-08-28 21:38:48.000000000","updated_by":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b667bb93a926ca8126b5b5d322e212b45fab17ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2024-08-28 20:43:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d3b50ca91e17be4bb6f1f160fae360b6c61ae58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 20:43:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"x/networking-cisco\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe problem appears in the the \"x/networking-cisco\" project stanza:\n\n  project:\n      name: x/networking-cisco\n      templates:\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5555","accounts_in_message":[],"_revision_number":1},{"id":"c74ca1a273146dae932e158e8e638b0cad9a4cce","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-08-28 20:44:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f6101872838abe96cb03a9376293181d73a3364","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-08-28 20:45:29.000000000","message":"Patch Set 1: -Code-Review\n\nOh, the error from Zuul points out that the tenant addition must merge first in a separate change.","accounts_in_message":[],"_revision_number":1},{"id":"4ceaadac67047e2da865d0fa56872e7863de7f4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"date":"2024-08-28 21:24:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6aed2051b043cf3612ab69c50f6b10bd261d6251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:34:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/241a7990c5a841e3aa5f8de7d5a18e79\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9a879a75ab1c46c7aa601b580100fdb8 : SUCCESS in 6m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eb7586e7cddd46e5840ac604713388cb : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"aa4635b8bd040ef6b20cb15e54edcdb3a40f7798","author":{"_account_id":32213,"name":"Sayali Vidyadhar Naval","email":"sayalinaval@gmail.com","username":"sayalinaval"},"date":"2024-08-28 21:45:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1cfb936518bc9a92860c8a4f5759ab972da1c5c5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-08-28 22:11:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a2a500c9d7731ad29ad3ce9440575d7590a9351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:11:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ef5c35138a54ced71b90d00cd5a8936bf8946e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:22:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f829f769494f048e0d1b701ac457f3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/90f49a3eb527495989776d54573def4a : SUCCESS in 6m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/788f93ce66824d88867f6326d6e7ed01 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6625e163a275ff93c16b09be453d729409a79058","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:22:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e9f8286fd60c12c1ca5df1672502e3760f57accc","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:29:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f30c951c39b44296baa6bc890fc077f0\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/1bbb7856879044b6aae24701ed8b28e0 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"683e663d279fc265332e466f792ac7369c8d7d22","revisions":{"8e97d7206b0c6d5cb85deeb495c773cec08e0952":{"kind":"REWORK","_number":1,"created":"2024-08-28 20:43:14.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/33/927433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/927433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/927433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/927433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/927433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/927433/1"}}},"commit":{"parents":[{"commit":"e94011bdaf28b06151fbff6a920471b1aa1ef583","subject":"Merge \"Retire unmaintained charm repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e94011bdaf28b06151fbff6a920471b1aa1ef583"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2024-08-28 20:39:30.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2024-08-28 20:42:42.000000000","tz":0},"subject":"Re-introduce gate jobs for networking-cisco","message":"Re-introduce gate jobs for networking-cisco\n\nThe networking-cisco project had been unmaintained, which led to it\nbeing removed from the gate (see [0],[1]). This patch re-introduces\nthe gate jobs for the project. Once merged, patches can then be\nintroduced to networking-cisco to resolve the failing gate jobs.\n\n[0]: https://review.opendev.org/c/openstack/project-config/+/923509\n[1]: https://review.opendev.org/c/openstack/project-config/+/925075\n\nChange-Id: Id88f3906a157a04b097b162950dedccc12a8fa90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e97d7206b0c6d5cb85deeb495c773cec08e0952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e97d7206b0c6d5cb85deeb495c773cec08e0952"}]},"branch":"refs/heads/master"},"683e663d279fc265332e466f792ac7369c8d7d22":{"kind":"REWORK","_number":2,"created":"2024-08-28 21:24:22.000000000","uploader":{"_account_id":18031,"name":"Thomas Bachman","email":"tbachman@yahoo.com","username":"bachmantech"},"ref":"refs/changes/33/927433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/927433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/927433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/927433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/927433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/927433/2"}}},"commit":{"parents":[{"commit":"e94011bdaf28b06151fbff6a920471b1aa1ef583","subject":"Merge \"Retire unmaintained charm repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e94011bdaf28b06151fbff6a920471b1aa1ef583"}]}],"author":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2024-08-28 20:39:30.000000000","tz":0},"committer":{"name":"Thomas Bachman","email":"tbachman@yahoo.com","date":"2024-08-28 21:22:03.000000000","tz":0},"subject":"Re-introduce gate jobs for networking-cisco","message":"Re-introduce gate jobs for networking-cisco\n\nThe networking-cisco project had been unmaintained, which led to it\nbeing removed from the gate (see [0],[1]). This patch re-introduces\nthe networking-cisco to the master list of projects. A subsqeuent\npatch is needed in order to re-enable the gate jobs.\n\n[0]: https://review.opendev.org/c/openstack/project-config/+/923509\n[1]: https://review.opendev.org/c/openstack/project-config/+/925075\n\nChange-Id: Id88f3906a157a04b097b162950dedccc12a8fa90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/683e663d279fc265332e466f792ac7369c8d7d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/683e663d279fc265332e466f792ac7369c8d7d22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
