)]}'
{"id":"openstack%2Fproject-config~928351","triplet_id":"openstack%2Fproject-config~master~I58d6b301c314c95b3e1a7abd0d4404a42f3e7e86","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I58d6b301c314c95b3e1a7abd0d4404a42f3e7e86","subject":"Add RackSpace Flex Nodepool dashboard in Grafana","status":"MERGED","created":"2024-09-06 12:45:53.000000000","updated":"2024-09-06 13:47:24.000000000","submitted":"2024-09-06 13:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928351","meta_rev_id":"aced13f6be9debc95a2b9055badcd070e09a36f1","_number":928351,"virtual_id_number":928351,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-06 13:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-06 13:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-06 13:17:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 13:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a09709f6f47b0871643c29f3a71118b32df52c2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-06 12:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ac63556445500f7320461e24994a8c84b10a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 13:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1289a3296c1c4fdab3c30b0a8debaf4e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7bcd5af935644e0b8f9ea4174cda6a39 : SUCCESS in 21m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5e9ab8895e114a9192e206defd46f4fd : SUCCESS in 6m 05s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/edcb5b562ae84cec8611f7cb99c208a1 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a94e5b3aa6356a2892e02b9c3c0aa0385177730f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-06 13:17:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nForwarding Clark\u0027s \"LGTM\" from IRC so we can get this added sooner and start to keep tabs on what\u0027s going on with the increased max-servers limit.","accounts_in_message":[],"_revision_number":1},{"id":"2f3838f7a4d6d0d814176113f4bf36c21feee493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 13:18:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce4686f027ff1b2fdf1331e0dd5f645fa5352c20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 13:45:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6e03336198b4be48c7bfd4c8582ff34\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/90d0e383699f4dc183f2952e1fb9e432 : SUCCESS in 16m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/974246b5e8a5411797ee935844cc3dcf : SUCCESS in 6m 44s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/661936d579a04bda91ee9019e48cc88f : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4db1fed6fb9e7f8aec81ba6d6529d1a9a4c04f9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 13:45:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aced13f6be9debc95a2b9055badcd070e09a36f1","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 13:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76bb3c9fb4524e2599fd0b71d9658fe6\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/f17a0a77864d4a83b39c0c7d37d0b3c3 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e7b15905bb72839b58ff5725c7f4c6a1c00165a","revisions":{"8e7b15905bb72839b58ff5725c7f4c6a1c00165a":{"kind":"REWORK","_number":1,"created":"2024-09-06 12:45:53.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/51/928351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/928351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/928351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/928351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/928351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/928351/1"}}},"commit":{"parents":[{"commit":"4398ebb1ccea89c024ed18e9558ad3657e2d516d","subject":"Merge \"Add default network for Rackspace Flex in nodepool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4398ebb1ccea89c024ed18e9558ad3657e2d516d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-09-06 12:43:59.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-09-06 12:43:59.000000000","tz":0},"subject":"Add RackSpace Flex Nodepool dashboard in Grafana","message":"Add RackSpace Flex Nodepool dashboard in Grafana\n\nWe\u0027re starting to increase utilization in this provider, so it\u0027ll be\nuseful to have easier ways to monitor things there.\n\nChange-Id: I58d6b301c314c95b3e1a7abd0d4404a42f3e7e86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e7b15905bb72839b58ff5725c7f4c6a1c00165a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e7b15905bb72839b58ff5725c7f4c6a1c00165a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
