)]}'
{"id":"openstack%2Fproject-config~928945","triplet_id":"openstack%2Fproject-config~master~Ieef4c95d81fe16caa5a1172f0e53c927c4807d60","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2024-09-11 22:51:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieef4c95d81fe16caa5a1172f0e53c927c4807d60","subject":"Add opendev/zuul-jobs repo","status":"MERGED","created":"2024-09-11 16:14:24.000000000","updated":"2024-09-11 22:55:48.000000000","submitted":"2024-09-11 22:51:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928945","meta_rev_id":"bfc9c2deaaa1117fa8083abac46ad417adf56be9","_number":928945,"virtual_id_number":928945,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-11 22:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-09-11 21:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-11 22:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-11 22:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-11 16:24:41.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-09-11 16:26:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 21:39:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"914bfb21f0ab777dfd5337ffe1670f3397eac615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-09-11 16:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8102684cd781ed1bbe5ec996a288a75433fb3849","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-11 16:24:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5f825a819a4d4f7b69a2215dbb389ec9a66e965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 16:26:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0296e41e71c24b0c85a75dd0387874a6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/abfaa972943f4d7b8f8d36621efced8f : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4f965db9c2ac46f99371e18cfa2ceb99 : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3dab606d9fe959ecfb3ca23fb36cd38df111e13d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-11 16:34:12.000000000","message":"Patch Set 1: -Code-Review\n\nOh, this needs an acl-file specified, or addition of a new ACL file matching the repository name.","accounts_in_message":[],"_revision_number":1},{"id":"896d297c03b9f5be15b4f8a71f854f4a6b4fac7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-09-11 17:48:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc14270eb4797d63c8cf853d4e316b45703cd8c2","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-09-11 17:48:56.000000000","message":"Patch Set 1:\n\nI suspect we will want to change it, but for now: system-config acls.","accounts_in_message":[],"_revision_number":1},{"id":"3638ead640c8ac26e11b61b6c0b59c85a027f9b4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-11 17:49:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c2e51c2caa4cc0fb1b71cbfa4addf04a4c8a1fc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 17:59:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cc1febd175a4a6d9db4a1f422e14518\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c2776416c2f44946b6c2c88bbb51bfae : SUCCESS in 5m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/288ad6f9c1fd415c84b8993f71523fa9 : FAILURE in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c269b3ef0a4379eac52e06f359ec74b8e2e53582","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-11 20:16:15.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"671dc088440469a8fbe29503bb5b676108089dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-09-11 21:25:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f6a29720f5f310f969a01ec18ac202d016f47d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:34:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf26b614f3294770a9841a3479f80c41\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/124134bca38b4c4bbc604850c2608193 : SUCCESS in 6m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/90a33ee79f6945ae8b332e4bb4aac8f0 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"62094e38f7461b6e8adb6c749da4673acff8c866","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-09-11 21:39:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aee51d8a35b9c38e6eecdb1a2a26e0c91ce1d56a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-11 22:45:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8f4b4164b2dd089ca3fe5e38e907aa72aaa1d76e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:45:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f2b6382bc2f36afa16de0be64c54cd68a13b5237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:51:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89feb5cf2c464040ac01cddcab181c5d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aa62dc4db7f542959a522aa40cc38049 : SUCCESS in 5m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/925e646ce1cd4b0f91a85cbc6446319b : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"44274b0cb114fc1549d60bb34671ebf11fe86aea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:51:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bfc9c2deaaa1117fa8083abac46ad417adf56be9","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:55:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b8fd48633342d2a2e8c568a4e036c1\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/d50c73fe758b4d29bc59ba75a4bc43ee : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"32c7f812bcc3d941ae091ce227d0ae150470ddc0","revisions":{"222b6652fcb61d18c2a107a6502e1e666f729f9b":{"kind":"REWORK","_number":1,"created":"2024-09-11 16:14:24.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/45/928945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/928945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/928945/1"}}},"commit":{"parents":[{"commit":"eb900c7c4d55c52ac5738e9436597110cd12bfa6","subject":"Merge \"Increase the boot timeout for Rackspace Flex nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb900c7c4d55c52ac5738e9436597110cd12bfa6"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-09-11 16:13:21.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-09-11 16:13:21.000000000","tz":-420},"subject":"Add opendev/zuul-jobs repo","message":"Add opendev/zuul-jobs repo\n\nThis is a repo to hold untrusted zuul jobs for the opendev tenant.\n\nChange-Id: Ieef4c95d81fe16caa5a1172f0e53c927c4807d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/222b6652fcb61d18c2a107a6502e1e666f729f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/222b6652fcb61d18c2a107a6502e1e666f729f9b"}]},"branch":"refs/heads/master"},"794ce561663e5991623680aca6f98ed5146667c8":{"kind":"REWORK","_number":2,"created":"2024-09-11 17:48:45.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/45/928945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/928945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/928945/2"}}},"commit":{"parents":[{"commit":"eb900c7c4d55c52ac5738e9436597110cd12bfa6","subject":"Merge \"Increase the boot timeout for Rackspace Flex nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb900c7c4d55c52ac5738e9436597110cd12bfa6"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-09-11 16:13:21.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-09-11 17:48:38.000000000","tz":-420},"subject":"Add opendev/zuul-jobs repo","message":"Add opendev/zuul-jobs repo\n\nThis is a repo to hold untrusted zuul jobs for the opendev tenant.\n\nChange-Id: Ieef4c95d81fe16caa5a1172f0e53c927c4807d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/794ce561663e5991623680aca6f98ed5146667c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/794ce561663e5991623680aca6f98ed5146667c8"}]},"branch":"refs/heads/master"},"32c7f812bcc3d941ae091ce227d0ae150470ddc0":{"kind":"REWORK","_number":3,"created":"2024-09-11 21:25:45.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/45/928945/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/45/928945/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/45/928945/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/45/928945/3"}}},"commit":{"parents":[{"commit":"eb900c7c4d55c52ac5738e9436597110cd12bfa6","subject":"Merge \"Increase the boot timeout for Rackspace Flex nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eb900c7c4d55c52ac5738e9436597110cd12bfa6"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-09-11 16:13:21.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-09-11 21:25:34.000000000","tz":-420},"subject":"Add opendev/zuul-jobs repo","message":"Add opendev/zuul-jobs repo\n\nThis is a repo to hold untrusted zuul jobs for the opendev tenant.\n\nChange-Id: Ieef4c95d81fe16caa5a1172f0e53c927c4807d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32c7f812bcc3d941ae091ce227d0ae150470ddc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32c7f812bcc3d941ae091ce227d0ae150470ddc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
