)]}'
{"id":"openstack%2Fproject-config~929776","triplet_id":"openstack%2Fproject-config~master~Ic6440dec8bfe91cf55679d7d7b8a03bc39e34197","project":"openstack/project-config","branch":"master","topic":"add-review-priority-for-neutron-stadium","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-09-19 06:38:52.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-20 15:38:23.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-09-18 19:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ic6440dec8bfe91cf55679d7d7b8a03bc39e34197","subject":"Add \u0027Review-Priority\u0027 for tap-as-a-service","status":"MERGED","created":"2024-09-18 13:46:56.000000000","updated":"2024-09-20 15:47:06.000000000","submitted":"2024-09-20 15:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929776-add-review-priority-for-neutron-stadium","meta_rev_id":"d10f0fd12ba04925b9035202724788237a3a21a3","_number":929776,"virtual_id_number":929776,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 15:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2024-09-19 06:38:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-09-18 19:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-20 15:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-20 15:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-18 13:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-18 18:08:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-09-18 18:08:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-09-20 15:25:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"49730cb872a851055af101431649ae184102dd44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-18 13:46:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16956855b009ef43aebff06cff5f12f5420c6450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 13:52:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ef7b993d09a42e49a15badb59a045dc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e28dffb5e989401fbd20520570e51982 : SUCCESS in 4m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/558d3e40f5274f0aafd6cf4bf8efa1c5 : FAILURE in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a6d61ba5b7bece21567f31b661c53bb4fbe9316b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-18 13:56:34.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dc05e7f5541b6ff9effada8b7679d92411981ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 14:03:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7cbcf5626c7408494d0c45228f5fd2b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c33f0e4211148db819ad962e9ec7950 : SUCCESS in 6m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8b6be4ddfbf04af985af1165c82c9f47 : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ad2ca49a0a7a45f1dd8eede1f952cefe204cd29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-18 14:12:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8c34b9d062115a25b346508ab79c2f1051396411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 14:26:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fae5e216f86462c9884715b8efe0712\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0894c76fabc8426bb201a88afd391612 : SUCCESS in 6m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4060b8372fc340498dc1e9041f002986 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"6aecb547720cbe4f7ce23df9d22d293ce17a3cc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-18 18:07:44.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0cbda6b623be8fe55a94b3d92c05884395ac9294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 18:19:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9a7c67fcd1e402aa98983e5a526580e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2afd03235f184de18ccc528fa7d8d0bd : SUCCESS in 5m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8cbf00af62d342bea5128a6bd350afa8 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":4},{"id":"42d99116308b4a8934e559eeeab2578dc8f9ffda","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-18 19:06:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6f1b8f21881156654ce556b626fd4d77bc10c92e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-09-19 06:38:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"431c9a71843276f7e1cd54aa5a95a486f990700a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-20 15:25:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4b457050a55f129b4a99ce55fde57eaba1113bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:26:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ac752dedeadd32e1d596afb788c69b235fa2a034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:38:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f55e7be67464c698f8bee394117b346\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/28918ef8eaa447f4b1e11f431c7983df : SUCCESS in 6m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/54f305f78b7c4e7fb38b1f58da49acfb : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":4},{"id":"d5da0764e4d357f066fe9a4f9bce06523071e54e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:38:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d10f0fd12ba04925b9035202724788237a3a21a3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:47:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34092f20c02946609d12c13a91649483\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/ed6d3fe20c194074a10dfeede0502493 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a327ac7fc8bb75a3de1f345989ccaf0d8c113f7b","revisions":{"efca6d6034d89f38775bdd0ec42d8479eade530f":{"kind":"REWORK","_number":1,"created":"2024-09-18 13:46:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/76/929776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/929776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/929776/1"}}},"commit":{"parents":[{"commit":"963bb0e3d189a8d51d6ad714a83287e129f49b72","subject":"Add nested virt labels to raxflex and openmetal providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963bb0e3d189a8d51d6ad714a83287e129f49b72"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-18 13:46:11.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-18 13:46:11.000000000","tz":120},"subject":"Add \u0027Review-Priority\u0027 for tap-as-a-service","message":"Add \u0027Review-Priority\u0027 for tap-as-a-service\n\nChange-Id: Ic6440dec8bfe91cf55679d7d7b8a03bc39e34197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efca6d6034d89f38775bdd0ec42d8479eade530f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efca6d6034d89f38775bdd0ec42d8479eade530f"}]},"branch":"refs/heads/master"},"9ba5897274d293849c5fb6bcf79060f586fec622":{"kind":"REWORK","_number":2,"created":"2024-09-18 13:56:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/76/929776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/929776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/929776/2"}}},"commit":{"parents":[{"commit":"963bb0e3d189a8d51d6ad714a83287e129f49b72","subject":"Add nested virt labels to raxflex and openmetal providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963bb0e3d189a8d51d6ad714a83287e129f49b72"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-18 13:46:11.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-09-18 13:56:34.000000000","tz":0},"subject":"Add \u0027Review-Priority\u0027 for tap-as-a-service","message":"Add \u0027Review-Priority\u0027 for tap-as-a-service\n\nChange-Id: Ic6440dec8bfe91cf55679d7d7b8a03bc39e34197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ba5897274d293849c5fb6bcf79060f586fec622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ba5897274d293849c5fb6bcf79060f586fec622"}]},"branch":"refs/heads/master"},"d7f68c1124a8daec69e5311f844551431a30bb7f":{"kind":"REWORK","_number":3,"created":"2024-09-18 14:12:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/76/929776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/929776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/929776/3"}}},"commit":{"parents":[{"commit":"963bb0e3d189a8d51d6ad714a83287e129f49b72","subject":"Add nested virt labels to raxflex and openmetal providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963bb0e3d189a8d51d6ad714a83287e129f49b72"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-18 13:46:11.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-18 14:11:53.000000000","tz":120},"subject":"Add \u0027Review-Priority\u0027 for tap-as-a-service","message":"Add \u0027Review-Priority\u0027 for tap-as-a-service\n\nChange-Id: Ic6440dec8bfe91cf55679d7d7b8a03bc39e34197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7f68c1124a8daec69e5311f844551431a30bb7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d7f68c1124a8daec69e5311f844551431a30bb7f"}]},"branch":"refs/heads/master"},"a327ac7fc8bb75a3de1f345989ccaf0d8c113f7b":{"kind":"REWORK","_number":4,"created":"2024-09-18 18:07:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/76/929776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/929776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/929776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/929776/4"}}},"commit":{"parents":[{"commit":"963bb0e3d189a8d51d6ad714a83287e129f49b72","subject":"Add nested virt labels to raxflex and openmetal providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963bb0e3d189a8d51d6ad714a83287e129f49b72"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-09-18 13:46:11.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-09-18 18:07:44.000000000","tz":0},"subject":"Add \u0027Review-Priority\u0027 for tap-as-a-service","message":"Add \u0027Review-Priority\u0027 for tap-as-a-service\n\nChange-Id: Ic6440dec8bfe91cf55679d7d7b8a03bc39e34197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a327ac7fc8bb75a3de1f345989ccaf0d8c113f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a327ac7fc8bb75a3de1f345989ccaf0d8c113f7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
