)]}'
{"id":"openstack%2Fproject-config~930029","triplet_id":"openstack%2Fproject-config~master~I8f170412e53ddc0fe52a3e2ca6d14ee27933ca96","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-09-20 15:34:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f170412e53ddc0fe52a3e2ca6d14ee27933ca96","subject":"Add official OpenStack jobs for whitebox-tempest-plugin","status":"MERGED","created":"2024-09-20 13:35:52.000000000","updated":"2024-09-22 02:28:43.000000000","submitted":"2024-09-20 15:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930029","meta_rev_id":"6a59ef4a0f019e252b70932b7a7b0d16e6f4f518","_number":930029,"virtual_id_number":930029,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 15:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-20 15:23:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-20 15:23:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-20 13:43:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-20 15:23:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"7a752c84b09c6395b64c6d7002dd9ad01b2079a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 13:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aece39d51f91b4249585637de65156f8ae0d09c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 13:43:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbfa2936fd60447cbec34cd5be0d0bb5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ccc75e69ac1745f19b9a67c73bd1350f : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8b69421381a126612f25df2236e95efcae1175cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 14:39:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae4e8046b725e4e4af8fc2798d71b24bd650c265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 14:47:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f6fb5ec203466c891ca13a1726b67b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/93808f551ff7417a86defeb231194d73 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4cc6f29f9aba38e1854c2f0c635f53bfb331c3cf","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-20 15:23:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ad832447a51635b44516470ad43e6c604e503eed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:23:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6d789b406a6ad5b7aae620bd367edb8a63d5fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:34:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7a9f33f0424ced85e6d3091d9d9c1f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/02bc471148a04de1b18ec3ef62411bf1 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7cd77957cb56ed4a0fe78215ed12bc1143f86432","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 15:34:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6fa75f085f83c929ce36dfc2a96a195034631ea1","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 16:47:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"324acd65f2500c87eaa4c04c17206bb33bb59ece","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-20 20:14:25.000000000","message":"Patch Set 2:\n\nThe mirroring job only runs when a change merges or a tag is pushed, so the next time one of those occurs for the repository it should update.","accounts_in_message":[],"_revision_number":2},{"id":"6a59ef4a0f019e252b70932b7a7b0d16e6f4f518","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-22 02:28:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe5709d7411c1a26259a6ad3da52eaa53a7e1231","revisions":{"2edc0ec1ebcb116c2f1d3a1b552aebb39d011f31":{"kind":"REWORK","_number":1,"created":"2024-09-20 13:35:52.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/29/930029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/930029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/930029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/930029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/930029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/930029/1"}}},"commit":{"parents":[{"commit":"963bb0e3d189a8d51d6ad714a83287e129f49b72","subject":"Add nested virt labels to raxflex and openmetal providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963bb0e3d189a8d51d6ad714a83287e129f49b72"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 13:35:00.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 13:35:51.000000000","tz":0},"subject":"Add official OpenStack jobs for whitebox-tempest-plugin","message":"Add official OpenStack jobs for whitebox-tempest-plugin\n\nThis should take care of mirroring to GitHub.\n\nChange-Id: I8f170412e53ddc0fe52a3e2ca6d14ee27933ca96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2edc0ec1ebcb116c2f1d3a1b552aebb39d011f31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2edc0ec1ebcb116c2f1d3a1b552aebb39d011f31"}]},"branch":"refs/heads/master"},"fe5709d7411c1a26259a6ad3da52eaa53a7e1231":{"kind":"REWORK","_number":2,"created":"2024-09-20 14:39:12.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/29/930029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/29/930029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/930029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/930029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/29/930029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/29/930029/2"}}},"commit":{"parents":[{"commit":"963bb0e3d189a8d51d6ad714a83287e129f49b72","subject":"Add nested virt labels to raxflex and openmetal providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963bb0e3d189a8d51d6ad714a83287e129f49b72"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 13:35:00.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 14:39:06.000000000","tz":0},"subject":"Add official OpenStack jobs for whitebox-tempest-plugin","message":"Add official OpenStack jobs for whitebox-tempest-plugin\n\nThis should take care of mirroring to GitHub.\n\nChange-Id: I8f170412e53ddc0fe52a3e2ca6d14ee27933ca96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe5709d7411c1a26259a6ad3da52eaa53a7e1231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe5709d7411c1a26259a6ad3da52eaa53a7e1231"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
