)]}'
{"id":"openstack%2Fproject-config~931999","triplet_id":"openstack%2Fproject-config~master~I591d99619ee41cd61627fe869e89bf52bcc2bde3","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2024-10-10 14:24:46.000000000","reason":"removed on reply"}},"hashtags":["opendev-niz"],"change_id":"I591d99619ee41cd61627fe869e89bf52bcc2bde3","subject":"Enable nodepool-in-zuul for opendev tenant","status":"MERGED","created":"2024-10-09 22:36:59.000000000","updated":"2024-10-10 14:37:00.000000000","submitted":"2024-10-10 14:30:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931999","meta_rev_id":"2d1f14d774ab4b8d38f567d832cd69736e343819","_number":931999,"virtual_id_number":931999,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 14:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2024-10-10 14:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-10 00:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2024-10-10 14:24:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 22:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 00:56:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"de385d51f3adefaf7f60d4d6322699cf68653d4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-10-09 22:36:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f3de263506a08e70435e867af0deb2cfd4cb44d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-10-09 22:42:55.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"78905cf7f11896348d42d99c24af67360442caca","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-10-09 22:45:06.000000000","message":"Hashtag added: opendev-niz","accounts_in_message":[],"_revision_number":2},{"id":"a8945aad244c37a137784745d1ed2c5c38b629d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 22:47:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d32c8e4975d24bd19fea058e614d9eda\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c0c15aef2f6745b2959536f827f0b2ea : SUCCESS in 3m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3bd023c1472247399ab312e7f30a346c : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ab3a8a819ba0cec70c9ca7540d302e226721f4e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-10-09 23:43:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"394b8d27cd9a3f73a235a34c5703826121e7d1b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 23:59:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476b0c5ae4f548a8a41b5d784f3d27e7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5e86b0122c6f4206a17b34df580f37bf : SUCCESS in 3m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4e79a433e7f54bc0902fc8ea31e968e0 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c9db1a7e0f992c1e3506bad36f1f62ae429ab39e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-10-10 00:56:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"54444d7d649a793299a284f03c6b81b47eda95f7","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-10-10 14:24:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ac2dd83dd33ebef62eb47c49f0ad1642b3b7c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 14:25:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2270c485274abaac4aa592c39c5e3c691af8ac41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 14:30:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/516fb25cebdf433687e5d1ae817e3c18\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/193572fff84645979ab226f42b6e69d9 : SUCCESS in 5m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a2b593b261e64de9ae4e55f63bff3ef3 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"eeb2419ed5c0d0565684050cd2377e4b19d3c489","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 14:30:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2d1f14d774ab4b8d38f567d832cd69736e343819","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 14:37:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1715fc0c7644e6afe5ab06b5abc62d\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/99ad13687c2d4f82be9211c31ea75e2c : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"441b0fba08356076072c985cf0fccfba883f5a78","revisions":{"1b49384127284cd9b731f1b5322fe38d3f570135":{"kind":"REWORK","_number":1,"created":"2024-10-09 22:36:59.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/99/931999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/931999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/931999/1"}}},"commit":{"parents":[{"commit":"a6f3515562ea4bf6058ac7b49f14c3b12b5e9d97","subject":"Use vexxhost/project-config in vexxhost tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6f3515562ea4bf6058ac7b49f14c3b12b5e9d97"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-10-09 22:34:50.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-10-09 22:36:38.000000000","tz":-420},"subject":"Enable nodepool-in-zuul for opendev tenant","message":"Enable nodepool-in-zuul for opendev tenant\n\nThis enables the experimental configuration objects on the opendev\nzuul-jobs repo.\n\nChange-Id: I591d99619ee41cd61627fe869e89bf52bcc2bde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1b49384127284cd9b731f1b5322fe38d3f570135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1b49384127284cd9b731f1b5322fe38d3f570135"}]},"branch":"refs/heads/master"},"fd72080f56b4e79aca65e7221b87aad64854bfba":{"kind":"NO_CHANGE","_number":2,"created":"2024-10-09 22:42:55.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/99/931999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/931999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/931999/2"}}},"commit":{"parents":[{"commit":"f3004706c169c35421da7d56a5ad8443ffd52adb","subject":"Use vexxhost/project-config in vexxhost tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3004706c169c35421da7d56a5ad8443ffd52adb"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-10-09 22:34:50.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-10-09 22:42:40.000000000","tz":-420},"subject":"Enable nodepool-in-zuul for opendev tenant","message":"Enable nodepool-in-zuul for opendev tenant\n\nThis enables the experimental configuration objects on the opendev\nzuul-jobs repo.\n\nChange-Id: I591d99619ee41cd61627fe869e89bf52bcc2bde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd72080f56b4e79aca65e7221b87aad64854bfba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd72080f56b4e79aca65e7221b87aad64854bfba"}]},"branch":"refs/heads/master"},"441b0fba08356076072c985cf0fccfba883f5a78":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-09 23:43:43.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/99/931999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/931999/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/931999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/931999/3"}}},"commit":{"parents":[{"commit":"8bc3ea0b92925f0ed6f06288a85ff92942503474","subject":"Use vexxhost/project-config in vexxhost tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8bc3ea0b92925f0ed6f06288a85ff92942503474"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-10-09 22:34:50.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-10-09 23:43:35.000000000","tz":-420},"subject":"Enable nodepool-in-zuul for opendev tenant","message":"Enable nodepool-in-zuul for opendev tenant\n\nThis enables the experimental configuration objects on the opendev\nzuul-jobs repo.\n\nChange-Id: I591d99619ee41cd61627fe869e89bf52bcc2bde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/441b0fba08356076072c985cf0fccfba883f5a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/441b0fba08356076072c985cf0fccfba883f5a78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
